)]}'
{"id":"openstack%2Fpuppet-octavia~849923","triplet_id":"openstack%2Fpuppet-octavia~master~Ibcd5596808085ecea19b42f953c01a252bf4f67c","project":"openstack/puppet-octavia","branch":"master","topic":"fix-test-desc","hashtags":[],"change_id":"Ibcd5596808085ecea19b42f953c01a252bf4f67c","subject":"Fix wrong test description","status":"MERGED","created":"2022-07-15 02:51:11.000000000","updated":"2022-07-15 04:00:34.000000000","submitted":"2022-07-15 04:00:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849923-fix-test-desc","meta_rev_id":"b76e523bb8aa0782a1a7c71ac7ee3e46e31f6149","_number":849923,"virtual_id_number":849923,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 04:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-15 03:40:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-07-15 03:40:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 03:18:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee706939b96968d90d88ec95a86a4df46fd9f2a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-15 02:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b86fc810c52762972de142e0e7abba1a5baa7bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 03:18:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63d26efc6a424a13b250c89ff0df0756\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/617955c3cb26416bb232f65739954d18 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22a8983a5e49480c934e11578ba836b7 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ba03e369334112bff10907d580337c : SUCCESS in 3m 40s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3214a8fc8491423594a6649980f8284a : SUCCESS in 16m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13bc593dbab547b890c2d38f1bce4d4f : SUCCESS in 10m 38s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5d33d83361f468785e7ba4151f0bc4d : SUCCESS in 10m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1dc0d9a358c473f90cd5df1600e7518 : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80bb1fb7ff4ac3ba5fc48f190f04fc57ea7f800d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-15 03:40:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65f8cf8ee86b16520026de7f5ed3ec4867ff5e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 03:41:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39f8206453e501f585f8a8cc44bbac634229507f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 04:00:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1be18c32f0b4d89921f0650224e8581\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf18f54cfdfb43c9be7f2e057f3d4a6c : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b66708bd2b7c48f6a383f12d909274cf : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52bfce1d60cd45f0bd54911c95c90c2b : SUCCESS in 3m 51s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f7faed258194424b338c630bf16a2b2 : SUCCESS in 17m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9016087b82ec4fa292b7087be38eb067 : SUCCESS in 11m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09ae043149a64d04b6b6dc5271f1c2d7 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b76e523bb8aa0782a1a7c71ac7ee3e46e31f6149","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 04:00:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c94a41a7fa670c2fee7e0497f2f34c5a21be988b","revisions":{"c94a41a7fa670c2fee7e0497f2f34c5a21be988b":{"kind":"REWORK","_number":1,"created":"2022-07-15 02:51:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/849923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/23/849923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/23/849923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/23/849923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/23/849923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/23/849923/1"}}},"commit":{"parents":[{"commit":"aa83e0b86fccb0572f8ca17785cc2e8542ff5d1c","subject":"Merge \"Fix typos in parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/aa83e0b86fccb0572f8ca17785cc2e8542ff5d1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-15 02:51:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-15 02:51:00.000000000","tz":540},"subject":"Fix wrong test description","message":"Fix wrong test description\n\nThe value with a white space should be accepted instead of being\nrejected. This change fixes the wrong test case description and makes\nsure the description explains what is actually tested.\n\nChange-Id: Ibcd5596808085ecea19b42f953c01a252bf4f67c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c94a41a7fa670c2fee7e0497f2f34c5a21be988b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c94a41a7fa670c2fee7e0497f2f34c5a21be988b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
