)]}'
{"id":"openstack%2Fpuppet-octavia~865088","triplet_id":"openstack%2Fpuppet-octavia~master~I13fb52a3f8e0e3237013da2c42c7b8baf322a006","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-25 17:40:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I13fb52a3f8e0e3237013da2c42c7b8baf322a006","subject":"Add support for [controller_worker] amphora_delete_* options","status":"MERGED","created":"2022-11-21 02:19:57.000000000","updated":"2022-11-25 17:41:57.000000000","submitted":"2022-11-25 17:40:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865088","meta_rev_id":"7311bf279a3f133c17ae629356fd829032672e68","_number":865088,"virtual_id_number":865088,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-25 17:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-11-21 15:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-25 13:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-25 13:59:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-21 03:55:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-21 15:21:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-25 13:59:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4e6b3940f21db5d7fd7c9a0add3fd2f7a943d219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 02:19:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec5089380f26eb17c508ca735aa7defe2148f81d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 03:06:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e83118cf758ad8c5c3ed0624a1bf6ed88dbe62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 03:55:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f93d2c8fe4a4e9e892361f6be7e498b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ac7d89b801494f8f92554d04e546b9 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbd322a2413b4e4ebcfbf0b74e72b927 : FAILURE in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee6bfa8f84f04042b708e928e140c9b4 : FAILURE in 2m 54s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/362d13b26a9d4e95b6c129cd4f89ecff : FAILURE in 12m 22s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e60cb36de36043cdb10962aa662b8a02 : FAILURE in 10m 05s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/163fc891803a4e679cf4456ad7e68ea1 : FAILURE in 8m 04s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63d9091dce0647fc8e14e5ebfcbc755a : FAILURE in 8m 01s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2858d4e98244bb6a3e419057c9b8396 : FAILURE in 6m 27s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e702f7171792416797e1f61ad0ba7beb : FAILURE in 9m 24s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ede879efd264de7aec43f629b489c7a : FAILURE in 5m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b6c355b93fd4315940b36d93bda9ed8 : FAILURE in 10m 33s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49105ec069024d148fa25b6f0463245c : FAILURE in 7m 15s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/546a26bc66e74770ac73338c81f4bbf1 : FAILURE in 9m 36s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85e6a1caaeb6436ca69b26c50eac0e51 : FAILURE in 5m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d9af98dcf494945aea909ff501a876d : FAILURE in 6m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcab771d9f684a5a98a953be3b102e6e : SUCCESS in 3m 59s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/571c331cbeca4e0ea65b39e327452f83 : FAILURE in 46m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a4a99571d3ad223ee7a2e41bfb0e795a9526603b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 04:50:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"91362c74080f58c524b14d2b2cdf05c3f6214e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 05:40:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a92988c2188b4767a9c65ec41e9a3612\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43167932c2684e04b7623edb6f46da56 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc26d764d97c47f598d3e49bf2065e32 : FAILURE in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38a0ced52bd74982b6b5c58d0898eb3f : FAILURE in 3m 11s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d231a656d7c4a8d9adf7d0971ff497d : FAILURE in 10m 40s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e62c109eb84f495f93acc25677aff84c : FAILURE in 7m 40s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91ead4b265a540559c97d15c6e0f84e9 : FAILURE in 8m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c8d0304b10e45c8a1976d1fa1fdca90 : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f29aad1891b4ce29dee1ec0276ec5a6 : FAILURE in 6m 32s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fe69359145547d5916a25dd06a89e93 : FAILURE in 10m 22s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43c65839084a47d381a666df05585e5e : FAILURE in 6m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b94212fbd82e4c1eb82284120ac06ec6 : FAILURE in 10m 33s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a234f5e5a73249ea9b2169424f5e93c2 : FAILURE in 6m 08s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/41485c33c1404ed49a6f884b6701983a : FAILURE in 9m 08s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd2dc815fbec45e1bc98fe987ab5f6d4 : FAILURE in 6m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f9b6849128b4a998bb0b7a38db420dd : FAILURE in 6m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6df11da54ba4a5cac5a12fc9119d64e : SUCCESS in 4m 20s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e71325cedf2b42a6a7886ad225219873 : FAILURE in 48m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ea0ba9d928b4fa23d5293215198f24ecb0f8867a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 05:58:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aea2a9019874e781591acd0d6415c86586533515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 06:54:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1147dada4d1a40c3bfcca70b39cac49d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9628b1bb22544119528a9eea0b21935 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae0c9218cde5448e80f8e0110684629c : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03e9b5aa548744e48048f13f7dfe6595 : SUCCESS in 3m 06s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c10313b1672431ca78b1708bdfb5970 : SUCCESS in 11m 34s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26884d70129c42e18ea8884647cf816b : SUCCESS in 7m 58s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad4b951955dc4199844bc82c1116464f : SUCCESS in 8m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02e31eff410f4703bd1e35ea9a20b1a7 : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc3c3fd68ba642258af9474dd2293f61 : SUCCESS in 38m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1858570903b6460da7caf1709b6349f3 : SUCCESS in 53m 06s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fb3783e636a4baca80bad569c6593d3 : SUCCESS in 39m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/671bf1bc7c974ba9800aa7de7c239151 : SUCCESS in 50m 43s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be27e5ffded34bc0b3521b12b19a3914 : SUCCESS in 18m 39s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8476942b59614ef1b88c645e26866555 : SUCCESS in 26m 20s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0df28439de345ff93ae70fdaf02d2d5 : SUCCESS in 20m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48c3bf73e97b40cf81ca4452ce2b198a : SUCCESS in 20m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbfbef118c43430f99ba1b3f1de6e487 : SUCCESS in 3m 01s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0b2b6da2fc50439f8bb06364b2f0f29e : FAILURE in 46m 26s","accounts_in_message":[],"_revision_number":4},{"id":"9c91b3d67c15d6580d45863b5e7c2f1e98aea53a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 06:58:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"de4254570071a85ef5bb30c61c05147cec072978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 08:00:50.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e2d21056da4448bb21d11c2a4f8684d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eedff14714c4177a1bc7b347d2462f7 : SUCCESS in 5m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc08615aac604330a7d7cc62e4e7bd10 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab5cfda4ed58410d87c092d2eb7c02d1 : SUCCESS in 4m 34s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/416da0d5e83546de84e52190b6551983 : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08a531b23d644cc6ad270cde4eaea9bd : SUCCESS in 9m 25s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eedc56d0eea4ddfa2058a97dc248202 : SUCCESS in 9m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1685034d64794d628f5dceb29a235643 : SUCCESS in 9m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea1d761812404a06a822d81ad23488d8 : SUCCESS in 40m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86e1a478e2244ed586f561f05984e82a : SUCCESS in 55m 58s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cc5950767ab4ec387c58acc2c78447d : SUCCESS in 40m 37s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dd0d9d9c416472896b0d77bbee406e3 : SUCCESS in 48m 47s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fbcfd4945b545a3908daff33045f375 : SUCCESS in 20m 18s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9074728ee1e343d4a46540522691faff : SUCCESS in 26m 14s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca7ad3528f5e4321a9ba33f455860855 : SUCCESS in 20m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f563a3c18bf4a8485c71c084db8adf2 : SUCCESS in 19m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89d0064bab194c96985d1384aedb3a6f : SUCCESS in 3m 02s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f94c872220354f008b34add3446e26ce : FAILURE in 59m 08s","accounts_in_message":[],"_revision_number":4},{"id":"329faffc4a6d28f2288340aa36691ac18e4bfdc8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-21 15:21:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ebc97220dc375c73976817e713601519f79ede1a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-25 13:59:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"41251c3c540d15c240135e35bfd9436cd728f6ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:51:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4e3d9a125c24e48b6aefd63def8702c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/204e7d68f6cd437c8caab3a14d5597a2 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8274e18eb884499ea9146280e8651111 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca263551ae4644219ec9c4f31913dcdb : SUCCESS in 2m 59s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/93caf1d75ff241cc97bbdaa70834a1fa : SUCCESS in 11m 18s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d74de67ee89a42b19ee4efa0415eafc6 : SUCCESS in 8m 25s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/130b195f8dfe42d59fbf30c6366f2231 : SUCCESS in 8m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aefe1774fa024c7786ab7fa2deb129e2 : SUCCESS in 8m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afe36f9bb62f40c7ac7139da22f73612 : FAILURE in 21m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d5f63494eab487fb2f7bbbd69ba2919 : SUCCESS in 54m 39s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2b4a38696f54d0fa93af4ac04895f4c : FAILURE in 19m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/167a72ef18c14fa986af89056c0a4d14 : SUCCESS in 55m 03s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/564db88f22534ba99f9176754cd4352a : FAILURE in 14m 27s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ce88b64e62e4322a6871a894abe9e03 : SUCCESS in 27m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00334865b4e4471cac2076b7a9dd8866 : FAILURE in 30m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77c7f902fe9341d492244edc21116354 : FAILURE in 28m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51a4a7fba2444b598179894089e46a05 : SUCCESS in 3m 13s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/727278e3d8f043438d9af46aa631aeed : SUCCESS in 1h 38m 44s","accounts_in_message":[],"_revision_number":4},{"id":"8de103be386904535f1d3a3ac16e7580f4679b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:53:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ddb6e209b65094fa2dbe9eed133bc3ac9edb6354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 17:40:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110c375e971f4b5db61b6a7e6409c963\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54591a0c416f42a7835d918789bcab62 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69c1b2bcdf504191b2aaf5ade058526b : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99beb6bc98b44a849f8030ab49c433de : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3208bcc00234e099f039b059d9d3b6d : SUCCESS in 12m 56s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e560ecd04a054b50bc2fddd165b75769 : SUCCESS in 7m 04s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f6f1525e363499e8ea384eb49105765 : SUCCESS in 8m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4208fdd66428423e81163a4cb6f1052a : SUCCESS in 52m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d5fdcde215b4a49a2e9c52b68dab103 : SUCCESS in 49m 55s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d72428c1dd014d79a9ff2aa27b578433 : SUCCESS in 24m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dfa2b1dd7e441349bef959ee9974bc2 : SUCCESS in 2m 47s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6eba6002597f4efeb7cc5c8b49472194 : SUCCESS in 1h 45m 19s","accounts_in_message":[],"_revision_number":4},{"id":"0fe6dea130ae49dd2ca1c5422ec4eb2d7d7df1bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 17:40:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7311bf279a3f133c17ae629356fd829032672e68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 17:41:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4826db3e7b1402094a954e451d0c6ff\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1010c90274104fd48f64d3cf5d7d7501 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6840de653f717ccb715d362dd4f054c8e57ec8ae","revisions":{"d09ee3f9b0dfeab2f2264e1e82d5c162d5c487a0":{"kind":"REWORK","_number":1,"created":"2022-11-21 02:19:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/865088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/88/865088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/1"}}},"commit":{"parents":[{"commit":"64869766d865584100b0ddd811f05b1887656bf6","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/64869766d865584100b0ddd811f05b1887656bf6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 02:18:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 02:18:36.000000000","tz":540},"subject":"Add support for [controller_worker] amphora_delete_* options","message":"Add support for [controller_worker] amphora_delete_* options\n\n... so that users can tune these options using the native class\nparameters.\n\nChange-Id: I13fb52a3f8e0e3237013da2c42c7b8baf322a006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d09ee3f9b0dfeab2f2264e1e82d5c162d5c487a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d09ee3f9b0dfeab2f2264e1e82d5c162d5c487a0"}]},"branch":"refs/heads/master"},"8190214e863d9f2a0c01203abc24b8b904639d15":{"kind":"REWORK","_number":2,"created":"2022-11-21 03:06:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/865088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/88/865088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/2"}}},"commit":{"parents":[{"commit":"64869766d865584100b0ddd811f05b1887656bf6","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/64869766d865584100b0ddd811f05b1887656bf6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 02:18:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 03:06:35.000000000","tz":540},"subject":"Add support for [controller_worker] amphora_delete_* options","message":"Add support for [controller_worker] amphora_delete_* options\n\n... so that users can tune these options using the native class\nparameters.\n\nChange-Id: I13fb52a3f8e0e3237013da2c42c7b8baf322a006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8190214e863d9f2a0c01203abc24b8b904639d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8190214e863d9f2a0c01203abc24b8b904639d15"}]},"branch":"refs/heads/master"},"81e298061c641d4bc20071780bf65e78df222fd2":{"kind":"REWORK","_number":3,"created":"2022-11-21 04:50:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/865088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/88/865088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/3"}}},"commit":{"parents":[{"commit":"64869766d865584100b0ddd811f05b1887656bf6","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/64869766d865584100b0ddd811f05b1887656bf6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 02:18:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 04:49:55.000000000","tz":540},"subject":"Add support for [controller_worker] amphora_delete_* options","message":"Add support for [controller_worker] amphora_delete_* options\n\n... so that users can tune these options using the native class\nparameters.\n\nChange-Id: I13fb52a3f8e0e3237013da2c42c7b8baf322a006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/81e298061c641d4bc20071780bf65e78df222fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/81e298061c641d4bc20071780bf65e78df222fd2"}]},"branch":"refs/heads/master"},"6840de653f717ccb715d362dd4f054c8e57ec8ae":{"kind":"REWORK","_number":4,"created":"2022-11-21 05:58:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/865088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/88/865088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/88/865088/4"}}},"commit":{"parents":[{"commit":"64869766d865584100b0ddd811f05b1887656bf6","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/64869766d865584100b0ddd811f05b1887656bf6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 02:18:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-21 05:58:05.000000000","tz":540},"subject":"Add support for [controller_worker] amphora_delete_* options","message":"Add support for [controller_worker] amphora_delete_* options\n\n... so that users can tune these options using the native class\nparameters.\n\nChange-Id: I13fb52a3f8e0e3237013da2c42c7b8baf322a006\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6840de653f717ccb715d362dd4f054c8e57ec8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6840de653f717ccb715d362dd4f054c8e57ec8ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
