)]}'
{"id":"openstack%2Fpuppet-octavia~867819","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~Id2d0efe4789cbbb60cb72b45b8dae7687985e4b3","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"oooci_teardown_8_wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-19 13:32:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2d0efe4789cbbb60cb72b45b8dae7687985e4b3","subject":"Remove centos8 wallaby tripleo jobs","status":"MERGED","created":"2022-12-15 15:28:40.000000000","updated":"2022-12-19 13:32:15.000000000","submitted":"2022-12-19 13:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867819-oooci_teardown_8_wallaby","meta_rev_id":"14512a633362a349b2d292cf5d930a2d68fec9e7","_number":867819,"virtual_id_number":867819,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 13:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2022-12-19 12:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-12-16 23:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-12-19 12:58:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 17:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-16 23:52:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-12-19 12:58:11.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"b19d177f729792f5c9b6120fce85416d20e86256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-15 15:28:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fcaa0f9be55a862a355779b943746476c47b254","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-15 15:37:32.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a0f44ccb4497a2170b47270781e4ced2b87cc837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 17:31:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1edb5fc249c0431eacac81c0d1683849\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/424828e0060343aab5f60ca7c0d56097 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82baab57f52942c6813c3af8c5cc5816 : SUCCESS in 4m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a507837528c414d942fd9a9e68b9a9c : SUCCESS in 11m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e88c7d336a7c466e8b7b2d18a32124d4 : SUCCESS in 7m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bab0d6db339405e9c2b689e9d90de72 : SUCCESS in 7m 00s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/faeab1f4161e4e2fa97f7b47eccc8bde : SUCCESS in 32m 17s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1daf9d3087434d98bb79d8224114e61a : FAILURE in 11m 58s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ccccea9a4e254f7b93651b1f2579473a : SUCCESS in 1h 41m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01fd01465547cf4d04856db97b371037664f2408","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-12-16 23:52:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9e08988d20c67948d5caba8700206b543074257a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-12-19 12:58:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e5a31fbdbfcdcfec98c43352507344afd791ad40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 12:59:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eae3ad72f0b9daeae5c852dcd2a1508a75d133c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 13:32:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73ba75d055247be89e122d55d1860c9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bf376a22a774a9785bb90c2f5440e20 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa8378c712054e1f949e27e24098804e : SUCCESS in 2m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdbecf4d92684e8a9714897c13a3f417 : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81fcb69e9ca647be8272ac86decfdb8d : SUCCESS in 7m 48s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/fad8570ecb6942d5850482d24c0c067c : SUCCESS in 28m 56s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a9b3bb2bcd241439f207f7d3f2eb052 : FAILURE in 13m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14512a633362a349b2d292cf5d930a2d68fec9e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 13:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92e7850a6378bb1755d8ef6a78a7ab3b47f4c037","revisions":{"f9ca7fcd5af893c1c137fa300d62b4d5facb2e31":{"kind":"REWORK","_number":1,"created":"2022-12-15 15:28:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/867819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/19/867819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/1"}}},"commit":{"parents":[{"commit":"35f7824b0b037427f7e976a5baa726f322c273e4","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/35f7824b0b037427f7e976a5baa726f322c273e4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-15 15:28:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-15 15:28:31.000000000","tz":540},"subject":"Remove centos8 wallaby jobs","message":"Remove centos8 wallaby jobs\n\nTripleO has decided to drop all centos8 wallaby jobs. This removes\nthe c8s tripleo jobs from Puppet OpenStack repos. Note that CentOS\nStream 9 is not officially supported by stable/wallaby release of\nour modules so the job is kept non-voting.\n\nChange-Id: Id2d0efe4789cbbb60cb72b45b8dae7687985e4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f9ca7fcd5af893c1c137fa300d62b4d5facb2e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f9ca7fcd5af893c1c137fa300d62b4d5facb2e31"}]},"branch":"refs/heads/stable/wallaby"},"92e7850a6378bb1755d8ef6a78a7ab3b47f4c037":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-15 15:37:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/867819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/19/867819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/19/867819/2"}}},"commit":{"parents":[{"commit":"35f7824b0b037427f7e976a5baa726f322c273e4","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/35f7824b0b037427f7e976a5baa726f322c273e4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-15 15:28:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-15 15:37:32.000000000","tz":0},"subject":"Remove centos8 wallaby tripleo jobs","message":"Remove centos8 wallaby tripleo jobs\n\nTripleO has decided to drop all centos8 wallaby jobs. This removes\nthe c8s tripleo jobs from Puppet OpenStack repos. Note that CentOS\nStream 9 is not officially supported by stable/wallaby release of\nour modules so the job is kept non-voting.\n\nChange-Id: Id2d0efe4789cbbb60cb72b45b8dae7687985e4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/92e7850a6378bb1755d8ef6a78a7ab3b47f4c037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/92e7850a6378bb1755d8ef6a78a7ab3b47f4c037"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[]}
