)]}'
{"id":"openstack%2Fpuppet-octavia~869616","triplet_id":"openstack%2Fpuppet-octavia~master~I0fadf39e316dd2d21da4930a92ee744ed729c725","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-17 01:20:45.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I0fadf39e316dd2d21da4930a92ee744ed729c725","subject":"Clean up deprecated controller class parameter","status":"MERGED","created":"2023-01-10 00:49:44.000000000","updated":"2023-01-17 04:52:52.000000000","submitted":"2023-01-17 04:51:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869616","meta_rev_id":"ca7a4aaf12e67d1589ec9965d83a1cc02f9cb56b","_number":869616,"virtual_id_number":869616,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-17 04:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-01-10 10:25:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-01-10 19:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-01-10 19:58:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 02:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 10:25:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-01-10 19:58:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"92f65b4c345a33ff8fd27af6d30fab7bfe10ad59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-10 00:49:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b60be21699ef3e78db70178a82d3a29aca626068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 02:29:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f86ef5a3ef444102836e745b9af5c7df\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9350df76b35e4868b13f575c1d5f0efd : SUCCESS in 2m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebe20e20dd88485faaf02135b191e43a : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98f0f605c9d24225b2d18183c1ef165e : SUCCESS in 2m 34s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ae8d15a18454b6e8acd473f8d666afd : SUCCESS in 9m 12s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/670ffd41213f4b07aabf50f5f46f02a8 : SUCCESS in 6m 42s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ffc2c5eb9f3436f8b80646939186299 : SUCCESS in 9m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24e467c4421148e6816889d59a25ac2c : SUCCESS in 6m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c134a3ea74584427b13b2b7d93fe293f : FAILURE in 30m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2011490259f46c4b54d7a4c49fe4461 : FAILURE in 44m 15s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/470068ef18ee465a9e30d40965bbbf28 : FAILURE in 31m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1059b2734e04e498e9ffe77ae2c6876 : SUCCESS in 56m 02s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e4478164c194f4280873142e131575f : SUCCESS in 20m 03s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7fe364c24cd4f65ac9263d8583db97d : SUCCESS in 23m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0017a7f1d3a4fe4a85ff5a70fee3038 : SUCCESS in 16m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf49674d68c943fca4cc1daf6a3d3e31 : SUCCESS in 13m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79d17c45301143818f3ab9e9cc23b980 : SUCCESS in 2m 39s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/96f5909af76e44af9e54aa9aff6359ea : SUCCESS in 1h 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d606f9d5f51769b9c1c5f3d3405ce113ba87d5b9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-10 03:34:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea47d8f7776a8176eafd3dbff2f0fed3e2f6280e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 05:06:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a40035193f944c4f89f132c0659368b3\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5308072166647e1b5f50f9ec37d0724 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7aab5a253afd4599a153db06069da934 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a98df174a06493c9540b607adedf074 : SUCCESS in 2m 34s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/7519155893c44ffebbbe5f83e985f5a6 : SUCCESS in 10m 38s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85461231da754a7cb0a5c363068d2f57 : SUCCESS in 9m 16s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/327a2ae318cb43f3a796679057603b15 : SUCCESS in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b88a368100334a9682e204ae7e08b31a : SUCCESS in 8m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89f44fdb7a0641bb8911220f32838ef5 : FAILURE in 30m 55s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dff19f7b75584eda88504c8785dc5a4e : SUCCESS in 51m 11s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/454a423923e341b6a478fe5656b99c7c : FAILURE in 26m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e15c40263d64d9ab909ddfca2d367b5 : SUCCESS in 51m 56s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3638ef665e174359a72d33de9906986b : SUCCESS in 20m 01s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/44aeb11165e645a7aa7f06bb05c9c827 : SUCCESS in 24m 41s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efced2b3b5474d68add3f27feadc8687 : SUCCESS in 20m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b5fc940fe644729938884d4edabc999 : SUCCESS in 17m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d229dd1db14f87b8d44e829f50dad8 : SUCCESS in 3m 50s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a396d5bd617a468791b3e5880c9fe13d : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3d81293fb9313b2528d028ee2ab0e45367bca7c3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-10 10:25:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d81d5d19fceea40f4c623ee7bd7cf9cc7e94973","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-10 19:58:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"521a902faecda028413e1b4f9013a1c12735c592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:59:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4fa076ba0409bd5d7bb454de41eed2abae55de3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 21:38:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aefa219c7f64ce4a7a47426ad13436b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da32d75808b44418a80f4b7f8ea866a0 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/049fa3404b344acdb6bce137e91eca46 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeb0406f0cb34830910c3cae8de8de0c : SUCCESS in 3m 35s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3be9e804d22544feb05ac2d4cd640f69 : SUCCESS in 12m 41s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02df780335854cc18140aa63b2ce8843 : SUCCESS in 9m 23s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d9aec225c114ef290e02e1f2074c370 : SUCCESS in 8m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/790cebed562244b890ed04bd8ffa5e46 : SUCCESS in 51m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef4545c8840948799dfefde181b46267 : FAILURE in 50m 19s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/880157a58662486798ef4e7a5023e308 : SUCCESS in 26m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e6f8104a024ba4b2d4eec56688103f : SUCCESS in 3m 58s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a6fe4f382f48415f9c1c73d40785bf71 : SUCCESS in 1h 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"14e2def17d036e3b01e3f60420185352575acf0a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-11 01:08:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"503c1de7e6a2ce1e2de6ac9030cbbb40507c2c77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 02:57:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3acc388fc32b4f89ae86b626797a3078\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b3eb06c39fc4561902e664b7ea7fd9e : SUCCESS in 2m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c6aa718e174435a967ae7cbc5397eb3 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a45d8cc56a1463fb0dcff67af895998 : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a541163ed9d747c1b1e3076182be35bc : SUCCESS in 12m 26s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7a844407ae84c0cbc907655129605b7 : SUCCESS in 8m 18s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e96b3512ee854d018f177286533d6a1c : SUCCESS in 8m 10s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd14ea04df0741468594fa2422c833c9 : SUCCESS in 8m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ca44b6ef19345159b9abcf2c37cd506 : FAILURE in 32m 55s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e34638a7df574043aadff29877db1576 : SUCCESS in 57m 11s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19ebe15871b942518ae6c08dc8523670 : FAILURE in 28m 54s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46ca764e15df4bc59dbf0a470fabe2c3 : SUCCESS in 50m 13s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6622873b88014d3cb97ea05e337ecb9a : SUCCESS in 18m 55s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8ee599d56f6471a80903d24d518fcb9 : SUCCESS in 22m 25s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31eda1337dbc4ad7a9f139ae1f65d53d : SUCCESS in 17m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75fe4f6e182941f6922e0e2ecaff9c03 : SUCCESS in 17m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b09665a000a4fe3b358c0b7b279e8b2 : SUCCESS in 3m 23s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cb6184ac90b64c9e90272ad8195e79b6 : FAILURE in 1h 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"637f6f2644d8567874b082514e20dce9e2443212","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-17 01:20:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da921b97c54b69cf38e138ca90c8dc50a606a4f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 03:16:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8b7ad2ab61c457bbf079d9db741ba12\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8abcb1cb64ce4d61aab714fc0c78cae1 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6f66687b194436eb5d66b41bd8147fb : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9ed17a2067d438190177902b2e061cf : SUCCESS in 11m 24s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f46346dd76484da289d0ef0cc4642487 : SUCCESS in 6m 59s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41d305587f184c7c97bdda749ea8fb5a : SUCCESS in 7m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/509e5dbfc24f4fd88884452f2830cf1f : SUCCESS in 40m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e112b855dd746208ec9b05246cdf5de : SUCCESS in 57m 09s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/beebeae0c02a43f6b13fc331867e41ce : FAILURE in 38m 46s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e14b162095c40979378cc4876d05800 : SUCCESS in 45m 41s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c3458a5abd4461aa9fcfec8ceaa1408 : SUCCESS in 26m 33s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21a512fcbd5443f989d8907dd5d377fe : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5f0213f79db4885af398985a3942479 : FAILURE in 6m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9250da16e0084bffa52ff214ed00bf40 : SUCCESS in 4m 08s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ce77a759d554f4cacc43f25c573622d : SUCCESS in 1h 39m 09s","accounts_in_message":[],"_revision_number":1},{"id":"18b03cbf07dfd1dea672de4ceb68c643c76cf935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 03:17:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6082246e5d18d1dbb60acaa352b34928aeefc5a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 04:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9525d7a364a8426197ce38401574f1fc\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28c9c0fc8fda441bb2e6a08063d78160 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49113c8599c04dc9afe27352fa08a15a : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/acb7121363964a3eaa4f359bcc7f8a76 : SUCCESS in 12m 28s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c711ef99a85941e2a6d44bf764c5dc16 : SUCCESS in 8m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/09513f882c764e3c9fe55402050792ba : SUCCESS in 51m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7b50f7885a145c084dbd657129802a5 : SUCCESS in 1h 01m 04s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ca2075ec3f740558577a7f25b3df93a : SUCCESS in 21m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2853bd4dd593487192e2e2786f271108 : SUCCESS in 2m 54s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8a6e28dc6a7c45a2b7f85e7dcf5a0769 : SUCCESS in 1h 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ba12ef32a9b83623a1a609c914167eefa91e63df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 04:51:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca7a4aaf12e67d1589ec9965d83a1cc02f9cb56b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 04:52:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d106445814b4e61a1d144b74226f6ef\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ce73d7e82de42959a94fc83b391c955 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db7f82c4e0a3217c37dea3b8cd766817c4bc45f0","revisions":{"db7f82c4e0a3217c37dea3b8cd766817c4bc45f0":{"kind":"REWORK","_number":1,"created":"2023-01-10 00:49:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/869616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/16/869616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/16/869616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/16/869616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/16/869616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/16/869616/1"}}},"commit":{"parents":[{"commit":"0c9b8a3828b666c297cbe48f5fa12cc88a0f6738","subject":"Merge \"Clean up deprecated health_manager parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0c9b8a3828b666c297cbe48f5fa12cc88a0f6738"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 00:48:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 00:49:35.000000000","tz":540},"subject":"Clean up deprecated controller class parameter","message":"Clean up deprecated controller class parameter\n\nThe port_detach_timeout parameter was deprecated in favor of\nthe separate networking class several cycles ago[1].\n\n[1] 4516e0cf9cbf22c78b07e7df6dda2f46afbd99ff\n\nChange-Id: I0fadf39e316dd2d21da4930a92ee744ed729c725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/db7f82c4e0a3217c37dea3b8cd766817c4bc45f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/db7f82c4e0a3217c37dea3b8cd766817c4bc45f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
