)]}'
{"id":"openstack%2Fpuppet-octavia~881157","triplet_id":"openstack%2Fpuppet-octavia~master~I0de1a0d3c76bfac55421832574a90be228d423ee","project":"openstack/puppet-octavia","branch":"master","topic":"log_file","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-28 04:05:15.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I0de1a0d3c76bfac55421832574a90be228d423ee","subject":"Stop overriding log_file","status":"MERGED","created":"2023-04-21 02:35:41.000000000","updated":"2023-04-28 05:50:28.000000000","submitted":"2023-04-28 05:49:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881157-log_file","meta_rev_id":"8809892f0321b391c05d63f25c806e6587c6dc92","_number":881157,"virtual_id_number":881157,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-28 05:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-04-24 15:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-04-25 05:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-04-25 05:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-21 03:43:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 15:45:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-04-25 05:51:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"38b66ba0606ac5470eaf333f86048dc4fa253625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-21 02:35:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ddfb1b0b563954aa94415caeffefe73591133b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-21 02:36:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"489268bd4d9cd90754bb29fd19f17251d7a18759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 03:43:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75027a5c5c7d4a97bea7322cb339c880\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fa969d9578c4a6197b4ecd762e7591a : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dea72797860c40b6be9a62fae9a2f85e : SUCCESS in 3m 10s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa590cfc3abf4e42821b55f3b436c253 : SUCCESS in 11m 05s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/017e0723840b424fbe7c17850554ba1a : SUCCESS in 7m 47s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddcbf39b08684109be2c5ce5cc9071ef : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adbca97ac6c04e12b4b1218a1a704c86 : SUCCESS in 40m 06s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5d913e79086467ea529c0b412cd1da8 : SUCCESS in 56m 28s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30a8ae1994434f9399102f92bdd5a833 : SUCCESS in 48m 51s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c64a60edcf034203abba34e776e02a36 : SUCCESS in 57m 40s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/fec64f94469e4d77a9aafeb4f142a2e5 : SUCCESS in 25m 48s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9da79e2de8644d26960bbe78bb118e1f : SUCCESS in 26m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5e89643101f4dd19f217d286ad398fe : SUCCESS in 20m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af10fcb19e454efdb75223a95021d557 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"cbf51bb4901a6c87579d797437edc91f8222c2ba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-04-24 15:45:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1d7d206ba020b21e17a509fb039683338857b635","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-25 05:51:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bd561e0fa0d2ce050dc0eaa8fb27603a036d74ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 05:52:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df5782913b8b7b2db4dc0028c45fac28692372d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 06:40:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b89e0dd3c844e51a0d2ca473c2308b1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6abbb3b42c140e3985d3545dcd2d27f : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfb0dd4d5e294f1ca46e57d5f0842670 : SUCCESS in 2m 29s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fc761ac5980438094e557dcae00a222 : FAILURE in 13m 30s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea84f90745fc477fa44add45aafe63cc : FAILURE in 7m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbf2ab09abeb4c43b416de5361ec5da6 : SUCCESS in 47m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/767e750062db43379ee5aa040f4ebfa4 : SUCCESS in 45m 02s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/81b81edee86047f5b2e625e1c7daddca : SUCCESS in 25m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06eab01027c246679b18f3960f402e57 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d0a3f7febe660a0a071620286688c902c6e3c61d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-26 13:08:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ce39326a1afca8e0fc3f1929234aefc202ee77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 14:14:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1205b63441ae49949f14811fdbfb4346\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb59447d6408433cae322ac058d35e9f : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c39539959c442c1bdf855d2cb609378 : SUCCESS in 2m 31s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2784d711085d4cbfbd96b9c72ca2266e : FAILURE in 10m 44s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b72d57e7592429e802b2b87ac08d9a1 : FAILURE in 7m 32s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4817b97caa554d4faacd8ca0b4d80eee : FAILURE in 6m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99857d4206224765abe4a3043086fde6 : SUCCESS in 45m 34s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/063ef0334f864d39a159e0028024480f : SUCCESS in 58m 33s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d655a81e08e448b0aa47ce3cd70db7f7 : SUCCESS in 46m 03s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffff9974f1d54eee8a19ffaeb14faf48 : SUCCESS in 1h 04m 04s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc449746d7a547089c49aed96da813cf : SUCCESS in 28m 00s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53561ee313fe493890577bbaac61562b : SUCCESS in 24m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50ed9ad61213431b9e16125578a22279 : FAILURE in 5m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca6a7da571c24e09ac291aaf2970f50d : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ffb0b3f68e0f0792acb2d1a206b01d4ab32ef071","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-27 00:39:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7eaed7e6fe92ea93dedba2a6e57a27f5481fba09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 01:34:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ce1a8e249f4e6a87bb49100d35ec9e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03714d7a18464bae91b539853749a58a : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/633cefbcdf134fcfacc387cbabc2f818 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a04dc85efe1e4d8da54394eb3802b777 : SUCCESS in 12m 05s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8119f6f3d7a340b6ac5140fe67cefff7 : SUCCESS in 8m 30s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c81c13ed45224e6e8e30ff4b1e34d28c : SUCCESS in 7m 28s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0e3b7f1b4ca48d48e84d5c3c6cadd04 : SUCCESS in 35m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b83413bd56a84e4b9ac60de657a2e7c2 : SUCCESS in 44m 07s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8262fd44bead43b6943f9c2427509b47 : SUCCESS in 43m 47s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f49784d32b84462aa43c05b2d018e605 : SUCCESS in 44m 20s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5704bd3bfa249429d8367a1aa6d39d4 : SUCCESS in 23m 30s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/122243d67f5f4b71bb5e6c4546e4037b : SUCCESS in 22m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e0bda27885148db9d17832e42b438c6 : SUCCESS in 19m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36bb658f6cfd4e4f8f77519fe8f7364e : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f1232487766c3f852ec07d11dc6c784956927cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 01:35:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"73b9406b0e1f69b6c3636dd5cd925617eca5dc26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 02:34:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1063b670c1aa4b81956f973abfb16c96\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f626b84ea34e9087bf9b3e3bf75544 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/338426b5559a41feb4ef30186fc75077 : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/54cfe13d30754093b14343ed07429d65 : SUCCESS in 12m 16s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76c5890455394d1493bf1ee085f59bd2 : SUCCESS in 9m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2905de82baca4189a7f1074ccb507298 : FAILURE in 57m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/73123464fa93406bb48bca7d1c0a0545 : SUCCESS in 58m 58s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/306a204554694f2d8dc1cd0568e06b23 : SUCCESS in 27m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f313d070f0942f69143445249b16ebc : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a7861e70319b34c2058cc42dbfac1d6bdbf2e2e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-27 03:24:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28d3e2dfb5bf771931ab3196234170ca60211a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 04:37:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b69f44465eb64207b9b339127573b945\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d040f8d03914b73a54b88ffbc761d91 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bb70338673a4507811db760fb9cc56b : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c72b1175dae4f9ca79096d6deabc7d1 : SUCCESS in 13m 46s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cffbdbb32f3d4a0eb357c41f0676a527 : SUCCESS in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b536690dc7544289bab6c941c9faf4d3 : SUCCESS in 8m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5050cf27aed4c1c95ebc805c489e3d1 : SUCCESS in 46m 59s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/45d413731607462497e925d2788b72fb : FAILURE in 55m 36s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2faa67bd547c4eb98cd087c95055c08b : SUCCESS in 49m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4846490e36fc4b5084fce889b45bc22a : SUCCESS in 51m 57s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d22d3953ac614533b33321514a9a4683 : SUCCESS in 23m 42s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd97945466e748948d294e969a99bf96 : SUCCESS in 18m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a79be5547404496d90d13b01913bbb9f : SUCCESS in 19m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e8c9e9956a4db38393d0f6058e3aae : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4b3d0e7916a3f575514c52ca3d63b500842eda8d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-27 06:30:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7058f3ea564f1c13fc725a4434734c8f909403d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:31:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b7eee81d41c406299daa68a7c8a25a6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fb8f8dca2c94b3287f35d1d2f2d63b9 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f20eae2299c40cd83270ab594c90888 : SUCCESS in 3m 19s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/75f63c7d29b1409f9bf5a4994d4516a5 : SUCCESS in 8m 55s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6634803f6796419c8197652b65fc39c9 : SUCCESS in 7m 24s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d3df3bd0c684ce296e5537c3cf8264f : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0da029ca9ee946999b784996d02a32a2 : SUCCESS in 38m 59s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ebff234cd934569b1b8e668eefc10f1 : FAILURE in 58m 15s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e4cab5c16ce4f39b36be441ed8cfac7 : SUCCESS in 41m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/747c02c4dd0d42f19bb4bed0674dc828 : SUCCESS in 59m 19s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc48bf4c0048420eb5677d04c80bdca4 : SUCCESS in 18m 41s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/980c69e77ee54f48adbd7f0754725970 : SUCCESS in 14m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc6abe4419814598ad97ee534b4c1ace : SUCCESS in 20m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7b3b6810af34504aec464151f15a01d : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"63c77aeb89571ae91ebbcc07cc1d8298af7945ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-27 09:44:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"271e6a56f74e6bad063352db03a7ea037ac2d771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:41:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61aaed32602644669d79feb2709382cd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf590929a5b84c66ab110825c2ae33ed : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7879296a9ad4431a850ff5a35ba3817d : SUCCESS in 3m 19s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/29f5667883764bcd869aae66bb8dc59b : SUCCESS in 13m 39s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77c3808a38c94f9c8cb6766989b308b1 : SUCCESS in 9m 27s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2f37f510b6449abaf178e31732afb55 : SUCCESS in 7m 27s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edb10aa871914babbf5dea082b0e8f75 : SUCCESS in 40m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/861049296b624f009f81d69c108a5ec2 : SUCCESS in 51m 43s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/524c88e1271d42aaab4caf315a1a8ce0 : SUCCESS in 44m 35s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/410ee3e390c84503879123377dfc7e69 : SUCCESS in 48m 04s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8164cf53ece7468781fe8e4b985d4390 : SUCCESS in 23m 41s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9db59b64aeb44f7592af7bf16fd3a6ea : SUCCESS in 20m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/137451d6cd5d4cd7822ca7a5501970cd : SUCCESS in 22m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc1055b8b43d40498ceb48676b7be217 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"00c61be512142a1e4306ec6d5ace985f71b3d6db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:42:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"015a69a77de34b35849e7176627ad9cd50f36696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 11:45:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/876755d186884f5e94ca655309e3d57a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cce61b7df57400c96656e870139bf97 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d834c3695e31485aa1acb1363fb15856 : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/620626a728a24852b96ef47b871822a7 : SUCCESS in 11m 34s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3145a252567640ecb7fbe84892899588 : SUCCESS in 8m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1640ac69e9b342c39342e4d16b1d2425 : FAILURE in 52m 01s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15f4b89e57964b23ada722819dfcb74c : SUCCESS in 56m 18s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4418dc1b20434e45864c34ab29a87910 : SUCCESS in 26m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0443baf7efdc4d4084b54f2d5afa7f71 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"716d56357f45b02ba500287fd793f488398abab3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-28 04:05:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d26e1f2f418d53ae22f93868c1863db99f2d223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 04:58:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa6baf787c1447f08376f4b1f974b909\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/754e28000f764321a6e9a7d3eb4322d2 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99d4a4fa9a8d4424a291fc038a9861c3 : SUCCESS in 2m 24s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e66d5eadef7143788cbfb8c96dadbac3 : SUCCESS in 9m 35s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3513ccc10b4469c8b11e32587a1beef : SUCCESS in 6m 43s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbb5ea2fcf8443b0902fa586df17e002 : SUCCESS in 7m 00s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c82e182476b4cb08a9aa51570aa6573 : SUCCESS in 36m 07s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1158ba01dcec4e079718d5b7d0557e83 : SUCCESS in 51m 00s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8109a56d9874e1cbc25c0ec07fdf0d2 : SUCCESS in 35m 49s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/658088fa4e5f41d8ad308dfc4784a2c7 : SUCCESS in 51m 25s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/15eeecdafd9149afbe1d620abfb8c27e : SUCCESS in 22m 54s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f016fca8abc14545bbb5d11474279021 : SUCCESS in 18m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b4e11a222ac48039bf58751749d8752 : SUCCESS in 18m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a168756f27054523a74a79bacf5aa1f6 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"14a091146d2abcb3003f0e59c97efd8762fb6a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 04:59:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce6533f9008cebdffce2266c610df5073eba9a23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 05:49:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67753711d0ce44ceb03a5e1a070545c2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/908114e4487a4f6d8689f7b851a5f183 : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9e7de974ed941a98e7bbcc36b193e1e : SUCCESS in 2m 21s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/96fe4305ec97411ab072d059e4c3a6ab : SUCCESS in 9m 45s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed4a54ce6a624e3daf87118558ea2fa0 : SUCCESS in 6m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44ca683caaf14e098387f1cdad54e44f : SUCCESS in 45m 15s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/704bab340e3047cda9a811a90bacc4e7 : SUCCESS in 48m 34s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/417b69cce80d4a97b1110128fdd084ba : SUCCESS in 20m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c347fd3c80749619b23f188d7c358ce : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4d9ca960a8208dda97ab395340a478ce77abfa90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 05:49:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8809892f0321b391c05d63f25c806e6587c6dc92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 05:50:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc5a2e3ae5e41f7a080e2ed3fa672a2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61ca1023749c442b9174fcc7a036cca6 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1586a07e5c5d5db16d065d770385074f2d8d6ad","revisions":{"3dcad552015f4f9fbd29ffe3b82326ca05b5a847":{"kind":"REWORK","_number":1,"created":"2023-04-21 02:35:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/881157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/881157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/1"}}},"commit":{"parents":[{"commit":"1b10a1bbe73052d8e50053f39616781b06d83da6","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1b10a1bbe73052d8e50053f39616781b06d83da6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-21 02:33:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-21 02:35:23.000000000","tz":540},"subject":"Stop overriding log_file","message":"Stop overriding log_file\n\nOctavia has multiple processes. So if we override log_file, multiple\nprocesses can write into the same file.\n\nChange-Id: I0de1a0d3c76bfac55421832574a90be228d423ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3dcad552015f4f9fbd29ffe3b82326ca05b5a847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3dcad552015f4f9fbd29ffe3b82326ca05b5a847"}]},"branch":"refs/heads/master"},"b1586a07e5c5d5db16d065d770385074f2d8d6ad":{"kind":"REWORK","_number":2,"created":"2023-04-21 02:36:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/881157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/881157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/881157/2"}}},"commit":{"parents":[{"commit":"1b10a1bbe73052d8e50053f39616781b06d83da6","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1b10a1bbe73052d8e50053f39616781b06d83da6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-21 02:33:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-21 02:36:43.000000000","tz":540},"subject":"Stop overriding log_file","message":"Stop overriding log_file\n\nOctavia has multiple processes. So if we override log_file, multiple\nprocesses can write into the same file.\n\nChange-Id: I0de1a0d3c76bfac55421832574a90be228d423ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b1586a07e5c5d5db16d065d770385074f2d8d6ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b1586a07e5c5d5db16d065d770385074f2d8d6ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
