)]}'
{"id":"openstack%2Fpuppet-octavia~883628","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fzed~I0d18ba7c5d832f7bd6485df721c8a94edecce6ad","project":"openstack/puppet-octavia","branch":"stable/zed","topic":"remove-tripleo-zed","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-22 04:41:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I0d18ba7c5d832f7bd6485df721c8a94edecce6ad","subject":"Zed-only: Remove TripleO job","status":"MERGED","created":"2023-05-19 04:47:47.000000000","updated":"2023-05-22 06:19:47.000000000","submitted":"2023-05-22 06:19:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883628-remove-tripleo-zed","meta_rev_id":"c4b52bd1c3138c898e09968b179159dec3448e0b","_number":883628,"virtual_id_number":883628,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-22 06:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":1,"date":"2023-05-19 10:17:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2023-05-22 03:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2023-05-22 03:37:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 05:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-19 10:17:17.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"}],"messages":[{"id":"1737af5fdae0b3b02cc4812828e836eabfb6c15e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-19 04:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b425e65ccd52c76751b03cc07d074ba24bddce38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 05:42:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748e095afd61486aba8ba3d01c40493d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a65e7b99424bdc865b05a117e6c75a : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55d59c029f23425d9a0e4af52f045241 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1af5df4fa0634b76a3ec6bfee9a7c7cc : SUCCESS in 2m 58s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b17643ad7424b0da6dd74b684256683 : SUCCESS in 9m 56s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/175ba7de4c064ec484f95f66e3dc3524 : SUCCESS in 9m 46s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f5a1a6320624006b6778a29bda1e3a5 : SUCCESS in 9m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/021cb447bb0442b48f63952093f253bc : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c82d80978ec4a4a90e9bd1b29185299 : FAILURE in 16m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc6afb9cc6d1494ea5bc4b95b287d9cc : SUCCESS in 48m 02s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3603c3d53fc47a3a62f501bdfb74bd8 : FAILURE in 16m 47s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02346cfa81914e658c7a0685882586a0 : SUCCESS in 48m 26s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae597ee9abc4e378ad996aed50317fe : FAILURE in 9m 59s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/76560eec4c1f4932ba7364a226c043d1 : SUCCESS in 26m 00s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dd9f32824df418ba78f3718066499ac : FAILURE in 27m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed7c4867e0a64f22b9eea1bcc4ce4f94 : FAILURE in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"138d875031656006fa90ef91b67b0ef24a40b7e5","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2023-05-19 10:17:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c9b659ad545b7f8234cd80f5abfab6b7ce36d87","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-22 03:37:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"729377e1a0bd99a7e382a270440a4e4863eb2131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 03:38:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90b7c0d53ddfd48658d92f58fb5f2f1314c51545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 04:39:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8748023fb854c3a81acfab96eb76966\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08ab44d7e98e4e4b9a6f80e66cff8c31 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/691a26f589304a83b30a548b2fc9de74 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cdc6242f7514721af086c2af415b7ca : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/32b5c6da87ea4715b5ddf8a70eb95ab2 : SUCCESS in 11m 01s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/027487331a3445b3a0eaae2d72193d78 : SUCCESS in 8m 08s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d05018f65a74d32859d49de36d009c8 : SUCCESS in 9m 01s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/949063dcae0545e6bf5e7f9da0918abf : SUCCESS in 57m 12s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a71eaaf1439f4edcbc28cce83d969185 : POST_FAILURE in 51m 36s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8eb2ae8fc4046e6af45e7133eb6cc09 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2853489bd53f40435a48012bfd360b1043f79656","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-22 04:41:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"842bd377c68723733bc19e8f6dc3a565d52e50bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 05:28:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d609ebe0be89492dad5b63b697a3044e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/031c4fbaa49b45d9ae72df61e59afaea : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9732e825886a48159f1b72eb198a73e4 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5af4401b02945c58b66062f7b9c01fc : SUCCESS in 3m 06s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/df4c6c470f584077a8944c36984f2653 : SUCCESS in 9m 22s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/786e6f4879864e36bf612eb771b7e039 : SUCCESS in 7m 13s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87680a54cc8d437eba7796b1e1c6bcf7 : SUCCESS in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3535c66d2ef48859a7d230955740c60 : SUCCESS in 7m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e69c46aa45054046b1bbd20eb0df11c2 : FAILURE in 19m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b18598054d34eed927ebc179a02d134 : SUCCESS in 44m 51s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/838961eb88274d49a4b3df9ced181ae4 : FAILURE in 15m 27s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23e870e8edcb492b9599d09a74bac795 : SUCCESS in 43m 55s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f44b3a19568f44b48df605eed92abf75 : FAILURE in 10m 27s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/21f28ed76c124fd5a88f48c31814a7b9 : SUCCESS in 26m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35e09d30e1db460ab0064f65ae94d6ee : FAILURE in 27m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b2866acf40445c2a1afba9fc54690e5 : FAILURE in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2c95c14d62edbf34f472de66f6fb1e56d8320de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 05:29:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9ee7e21ecfe8b47f208ad0a1460e4554760265f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 06:19:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49cb8560141f4a39b0abc45fe43496cf\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0906896431446528468ca260472abc8 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb52a8287dcb4209b7d0be69d7611000 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85a1954248f3441e9596c37eb1c5d124 : SUCCESS in 3m 06s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d158649cb9ec41ebb543d15a613ab935 : SUCCESS in 10m 43s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/681b4a543ee744399b2c193aced02f4f : SUCCESS in 7m 23s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0139cb9952847bca0cd4900f714eb7a : SUCCESS in 9m 19s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23a4dfdd78844ac69bf5bfd184bb0cf6 : SUCCESS in 50m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/99554b6e17c940e78fd83cdb1fa3c5c0 : SUCCESS in 44m 03s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bddf7b9342c40a79a76ebce129f0627 : SUCCESS in 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c4b52bd1c3138c898e09968b179159dec3448e0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 06:19:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34e0aea22dac1fe9d064f05d061220e96e6af0b8","revisions":{"34e0aea22dac1fe9d064f05d061220e96e6af0b8":{"kind":"REWORK","_number":1,"created":"2023-05-19 04:47:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/883628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/28/883628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/883628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/883628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/28/883628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/28/883628/1"}}},"commit":{"parents":[{"commit":"38655695f276a69e8e6e487a028008f28c56c1f4","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/38655695f276a69e8e6e487a028008f28c56c1f4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-19 04:47:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-05-19 04:47:38.000000000","tz":540},"subject":"Zed-only: Remove TripleO job","message":"Zed-only: Remove TripleO job\n\n... because the stable/zed branch of TripleO repos is no longer\nmaintained.\n\nChange-Id: I0d18ba7c5d832f7bd6485df721c8a94edecce6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/34e0aea22dac1fe9d064f05d061220e96e6af0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/34e0aea22dac1fe9d064f05d061220e96e6af0b8"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
