)]}'
{"id":"openstack%2Fpuppet-octavia~905439","triplet_id":"openstack%2Fpuppet-octavia~master~I3f4cf18b70420a509ad971fea32277a7a9b59dc3","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-12 17:43:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f4cf18b70420a509ad971fea32277a7a9b59dc3","subject":"Restrict access to certificate files","status":"MERGED","created":"2024-01-12 13:33:32.000000000","updated":"2024-01-12 17:43:40.000000000","submitted":"2024-01-12 17:43:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905439","meta_rev_id":"ca0669c99db032edda0f6d1924f6c9e2f5a77bc9","_number":905439,"virtual_id_number":905439,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 17:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-01-12 16:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-01-12 16:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-01-12 16:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 14:38:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 16:19:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-01-12 16:33:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ec14b7a895e8c4d11a36bfe837a4a1ac222d11d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-12 13:33:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"733dee8430758b889d73e0aa869d7c7ccaf835de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-12 13:34:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cf41af57915f961bff6ce4d6f2590029481e0c73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-12 13:36:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b2cacc89b542016e074d075df411d7e27a29343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 14:38:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bec075ba4844965a9478bc47c814ba2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b03151eb11d4b249f56cad6d12c69e9 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76c84061354348728c7b0fa2998d95ab : SUCCESS in 3m 00s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf61bf4d433b460e8e5217b1fefbc80f : SUCCESS in 5m 28s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/746d2b7eb88349de883d8349f6ff5120 : SUCCESS in 10m 35s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0640c306e5341628778f9928c673120 : SUCCESS in 9m 59s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/c003051864ab420a9e392022c291e6aa : SUCCESS in 12m 30s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ee52a3a47ae43b79cbd2c837350add7 : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03f7e56aab864ccb93703cb35f5b8c85 : SUCCESS in 39m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9961b2113314cc780c00f3393858737 : SUCCESS in 54m 24s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0efb12c1e3024467a154ba23cdd8df29 : SUCCESS in 45m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44497db62a284fba89f2061efd4cf709 : SUCCESS in 54m 31s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/856f590afab34012b68e7ea8b64b3d07 : SUCCESS in 26m 06s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9db7181c7154fd39584e0bbbb275dff : SUCCESS in 18m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18057f76ad39454aa9020821b71b6dc6 : SUCCESS in 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"588ff282e47a239e3020870531dd07830bffca59","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-12 16:19:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"91975960a1a9c189ff29651fa233e7b95eee85e2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-12 16:21:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b7ad0b316fa5c2fe46538304972082cf8def41c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-12 16:29:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f0809f4078fb10babd6362cc92df6ed4583288e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-01-12 16:33:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aab4ff2aa9b2ed3772b55a427e974cdad458503e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 16:34:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d6fc3a6d1f6376b414adacc769890cabc3802daf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:43:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523a6e7199e44523ac72ca63b18267d5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96f45628770c43aa9110503effaa22be : SUCCESS in 4m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed27e276842d41ab80621c9e0cff7ad2 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b3a3836ece34c75b4e80afb37d32bdb : SUCCESS in 11m 42s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/995d5b0fbd7e4c2aa820afaa6d05f457 : SUCCESS in 8m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1741ad6214d48aab3867837c5b10561 : SUCCESS in 59m 34s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a164030f867462b98e0eda125a7effd : SUCCESS in 49m 27s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a8ea880ba7042e0acb3844273f216ff : SUCCESS in 25m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ca0669c99db032edda0f6d1924f6c9e2f5a77bc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:43:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"76e1ac5e8c6f700025a849f10c555cdd686f822d","revisions":{"27c6a7958fcd5fb99d92572da21275febf332914":{"kind":"REWORK","_number":1,"created":"2024-01-12 13:33:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/905439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/39/905439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/1"}}},"commit":{"parents":[{"commit":"869993e13a12c073ebc0b823b8d9958c37dd5e57","subject":"Merge \"Remove unused octavia-db-sync-service tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/869993e13a12c073ebc0b823b8d9958c37dd5e57"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 13:31:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 13:32:48.000000000","tz":540},"subject":"Restrict access to certificate files","message":"Restrict access to certificate files\n\nThe certificate files don\u0027t need x bits. Also these files, especially\nthe private key file should have very restricted access.\n\nChange-Id: I3f4cf18b70420a509ad971fea32277a7a9b59dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/27c6a7958fcd5fb99d92572da21275febf332914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/27c6a7958fcd5fb99d92572da21275febf332914"}]},"branch":"refs/heads/master"},"a382404ddd077506698fa1a9bf55ce2bc1f927f6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-12 13:34:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/905439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/39/905439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/2"}}},"commit":{"parents":[{"commit":"869993e13a12c073ebc0b823b8d9958c37dd5e57","subject":"Merge \"Remove unused octavia-db-sync-service tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/869993e13a12c073ebc0b823b8d9958c37dd5e57"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 13:31:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 13:33:56.000000000","tz":540},"subject":"Restrict access to certificate files","message":"Restrict access to certificate files\n\nThe certificate files don\u0027t need x bits. Also these files, especially\nthe private key file should have very restricted access.\n\nCloses-Bug: #2049203\nChange-Id: I3f4cf18b70420a509ad971fea32277a7a9b59dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a382404ddd077506698fa1a9bf55ce2bc1f927f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a382404ddd077506698fa1a9bf55ce2bc1f927f6"}]},"branch":"refs/heads/master"},"76e1ac5e8c6f700025a849f10c555cdd686f822d":{"kind":"REWORK","_number":3,"created":"2024-01-12 13:36:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/905439/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/39/905439/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/39/905439/3"}}},"commit":{"parents":[{"commit":"869993e13a12c073ebc0b823b8d9958c37dd5e57","subject":"Merge \"Remove unused octavia-db-sync-service tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/869993e13a12c073ebc0b823b8d9958c37dd5e57"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 13:31:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 13:35:50.000000000","tz":540},"subject":"Restrict access to certificate files","message":"Restrict access to certificate files\n\nThe certificate files don\u0027t need x bits. Also these files, especially\nthe private key file should have very restricted access.\n\nCloses-Bug: #2049203\nChange-Id: I3f4cf18b70420a509ad971fea32277a7a9b59dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/76e1ac5e8c6f700025a849f10c555cdd686f822d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/76e1ac5e8c6f700025a849f10c555cdd686f822d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
