)]}'
{"id":"openstack%2Fpuppet-octavia~923693","triplet_id":"openstack%2Fpuppet-octavia~master~Idc997403770c99be654419b4d0d4569e8669c2fb","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-29 08:36:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idc997403770c99be654419b4d0d4569e8669c2fb","subject":"Support options for authentication and SSL for Redis Sentinel","status":"MERGED","created":"2024-07-09 02:53:39.000000000","updated":"2024-07-29 08:37:26.000000000","submitted":"2024-07-29 08:36:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923693","meta_rev_id":"f33c82b91af3f09f3e019268df4b5cdbccb50e37","_number":923693,"virtual_id_number":923693,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-29 08:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-07-10 21:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-07-29 07:29:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-07-29 07:29:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-09 03:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-10 21:35:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-07-29 07:29:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"19a63ab435a13ab28fd563e59692a101a6f6da2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-09 02:53:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33fafec27716a69b51d34c8a7817d29a0f43c850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 03:54:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39135ff3d46b443dbeff9e9775819cb4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a62b66199d401aa12cf3a5229cdf3c : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c2d3f9fe736487985b2e71e96d5e2a7 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/09f4b16ed35d4721909813f86d97caad : SUCCESS in 4m 39s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a27ed69dead42a498bc4f2098529e36 : FAILURE in 11m 10s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82e8cae3f4e9463e8ae408a04ed3562d : FAILURE in 7m 29s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/05cf87d405a94596b96092fcd3d12756 : FAILURE in 10m 27s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbdf5d663fd043fbb8c6696f0d4150d7 : FAILURE in 6m 58s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81bbbddb573747edb74d6bae30365513 : SUCCESS in 45m 32s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/618c307667ab4c39aa6ae30d369db920 : SUCCESS in 51m 23s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b1150c29d754d8589ce7487f0bdddab : SUCCESS in 36m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a0f8c7d481d48a9bb3f43534db74448 : SUCCESS in 53m 56s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/de9e5d4e56a0461392c84a6884235c8f : SUCCESS in 24m 22s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/720c0b26cdb0444c99ad885245c9710b : SUCCESS in 16m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/843bd859c2494fa196aabad5354128de : SUCCESS in 16m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c1dbdd0e6084818bf878460cece849c : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6d27ebc461f02a0d256b80c171da4d6e4089cc6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-09 04:21:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"855b68c9f5f31dc2f0c68acc9a81322ab5803a1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-09 04:41:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c312457e02d28ec54bdef15678f064e380b3c9bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 05:44:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe07b87d0d943e7847ba1f7e08d2f3e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/037b145a3f074f01a8ba87947eed8c4a : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c9f49f4572a45b0947cded7ef74c63b : SUCCESS in 2m 57s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b915488d82354a299dcfb923103dfd48 : SUCCESS in 5m 06s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1264e30706b4588bd45993e0c21f528 : SUCCESS in 13m 13s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad99971041b0425f9b66aba41b5ceef5 : SUCCESS in 7m 32s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/d48ff4cf7b96421eb7c755e42cc38d79 : SUCCESS in 13m 14s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ddd6d30526a4f57ab558efa117cae74 : SUCCESS in 7m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c82a1474eb64e51a18021da8a8f8581 : SUCCESS in 41m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f204d3667cd4fdea585808a8d0c3735 : SUCCESS in 1h 00m 45s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61a37a04888345c58f3d3ef574282d0a : SUCCESS in 45m 26s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b031b23a5f44ee78c4a1ed4347ef9be : SUCCESS in 45m 48s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c468dd6ca29468f9be918e339806f4e : SUCCESS in 21m 29s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f85fcf6f38c4a688fe2f04bdddcef90 : SUCCESS in 14m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/661c8cc85ad14a67b1097d75c7f39855 : SUCCESS in 14m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac616a08937d478fa6351e4876725611 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"04f898d0803d260cbea8c101f5f887042f41d358","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-10 21:35:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1bf79b01dc85b52796cbad1a020da54bc093c67b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-29 07:29:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b9d812f5710ced5feaf5a6fc1eb74aa0ad709707","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 07:30:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa39e50b706fb47b52af467a3b363f2ce072444f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 08:36:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a81de116cf4483f87d3736957e998d0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99ffcc5ad11544b9bf54e321d0625e8a : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87236d3b053541ab857a420b5ada7c78 : SUCCESS in 2m 28s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/94cf4a80106b405486a9c373a3a33478 : SUCCESS in 10m 00s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1ee45d29ab141378984f031b2857e3a : SUCCESS in 7m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/21a72156195647e39c77cd903bf9e1e7 : SUCCESS in 1h 00m 56s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/641d32b2709a486695e91e47b4e4e0b5 : SUCCESS in 54m 33s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/de9182279e334882b3cec33c4457ed07 : SUCCESS in 27m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d18ae2515fe476a9cc9c2a8bc2a21a6 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":3},{"id":"8537b574e05a562914f9b43ae5b39b237bebe1a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 08:36:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f33c82b91af3f09f3e019268df4b5cdbccb50e37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 08:37:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61154b437a034e4e84433cc8297133e1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1039baaf6bf84b2cb5e4e98a76eb9111 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"058c922576b0e52905aaf27ec2110d7cdf566a6c","revisions":{"4b06581513848a30245df5e00dbaa67127734ec4":{"kind":"REWORK","_number":1,"created":"2024-07-09 02:53:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/923693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/923693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/1"}}},"commit":{"parents":[{"commit":"6b1dd984dc001f4ddeffefc6f8a57beebcefeffb","subject":"Merge \"Remove support for amqp1 messaging driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6b1dd984dc001f4ddeffefc6f8a57beebcefeffb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-09 02:51:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-09 02:53:27.000000000","tz":540},"subject":"Support options for authentication and SSL for Redis Sentinel","message":"Support options for authentication and SSL for Redis Sentinel\n\nAdd support for the new options to enable authentication or SSL for\nconnections to Redis Sentile.\n\nDepeneds-on: https://review.opendev.org/912355\nChange-Id: Idc997403770c99be654419b4d0d4569e8669c2fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4b06581513848a30245df5e00dbaa67127734ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4b06581513848a30245df5e00dbaa67127734ec4"}]},"branch":"refs/heads/master"},"d8ebbd4f4b2a417cc5845bd131744fa934a48d59":{"kind":"REWORK","_number":2,"created":"2024-07-09 04:21:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/923693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/923693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/2"}}},"commit":{"parents":[{"commit":"6b1dd984dc001f4ddeffefc6f8a57beebcefeffb","subject":"Merge \"Remove support for amqp1 messaging driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6b1dd984dc001f4ddeffefc6f8a57beebcefeffb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-09 02:51:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-09 04:21:36.000000000","tz":540},"subject":"Support options for authentication and SSL for Redis Sentinel","message":"Support options for authentication and SSL for Redis Sentinel\n\nAdd support for the new options to enable authentication or SSL for\nconnections to Redis Sentile.\n\nDepeneds-on: https://review.opendev.org/912355\nChange-Id: Idc997403770c99be654419b4d0d4569e8669c2fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d8ebbd4f4b2a417cc5845bd131744fa934a48d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d8ebbd4f4b2a417cc5845bd131744fa934a48d59"}]},"branch":"refs/heads/master"},"058c922576b0e52905aaf27ec2110d7cdf566a6c":{"kind":"REWORK","_number":3,"created":"2024-07-09 04:41:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/923693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/923693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/923693/3"}}},"commit":{"parents":[{"commit":"6b1dd984dc001f4ddeffefc6f8a57beebcefeffb","subject":"Merge \"Remove support for amqp1 messaging driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6b1dd984dc001f4ddeffefc6f8a57beebcefeffb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-09 02:51:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-09 04:41:21.000000000","tz":540},"subject":"Support options for authentication and SSL for Redis Sentinel","message":"Support options for authentication and SSL for Redis Sentinel\n\nAdd support for the new options to enable authentication or SSL for\nconnections to Redis Sentinel.\n\nDepeneds-on: https://review.opendev.org/912355\nChange-Id: Idc997403770c99be654419b4d0d4569e8669c2fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/058c922576b0e52905aaf27ec2110d7cdf566a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/058c922576b0e52905aaf27ec2110d7cdf566a6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
