)]}'
{"id":"openstack%2Fpuppet-octavia~933490","triplet_id":"openstack%2Fpuppet-octavia~master~Ib7bcf34cb74135f5ff5f9ad8ba77d707e493f24a","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-29 04:26:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ib7bcf34cb74135f5ff5f9ad8ba77d707e493f24a","subject":"Add support for [task_flow] jobboard_redis_backend_db","status":"MERGED","created":"2024-10-26 12:53:30.000000000","updated":"2024-10-29 07:27:35.000000000","submitted":"2024-10-29 07:26:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933490","meta_rev_id":"1f77021b0c9e9a102a926f6e2e78440404841e28","_number":933490,"virtual_id_number":933490,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 07:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-10-28 07:14:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-10-28 14:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-28 14:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 13:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 07:14:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-28 14:12:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"02428bd6e22bd7335b3a4fa4bb46100219b958ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 12:53:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48c97a1c3b7a59c6393496aded3f30eaf5597b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 13:08:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0cb8b8fd64f4c318d325cfa89cce794\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40ad0960083f464ca895235ed8c0a65f : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8d743a9ad934a698dbc00a2a8d72bd8 : FAILURE in 2m 31s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1f89d2f14724bcca33f2114cad5040b : FAILURE in 3m 07s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/94aa325ac2e143e5913697b9963b22ac : FAILURE in 9m 39s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdcb8006e4f34944a1874664a6426b5c : FAILURE in 5m 47s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cc6fa42ef514276b71842bb8cce302c : FAILURE in 7m 26s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95eb669fe8774f4d9997b92b1a2c32c2 : FAILURE in 5m 21s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07a81e3f0bae44188829f8a4b034df5b : FAILURE in 4m 32s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68d29a26edbc418eada86af8e13ce2c2 : FAILURE in 7m 48s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8c1a645ad904d8a8f1af6a83c885056 : FAILURE in 4m 35s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c12668728ca4b8f9e173a12ccaa6289 : FAILURE in 7m 49s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fa84e341eed448a95d6c4217e250772 : FAILURE in 6m 13s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1d34d0d2c714da89756417c78c3b657 : FAILURE in 3m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7163b807734e46699201f700ac163c2a : FAILURE in 4m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27755dee62cf4c7ba61bac296b0ff032 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"695acf449aef69f833197db010472b3dcb2637c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 13:25:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ca61a1e955d1fda693470231b0668bb1cbdbbee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 14:37:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fe8ee702a024493b9bd39f56e9211be\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59693c4f54c1451f82fc3f5cf014ebd9 : SUCCESS in 2m 06s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7de6dbd29b524f91a7d56641bc428b54 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f3563adc1ce46e5a39de41a40e218ff : SUCCESS in 4m 08s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b3081c7aab849198f0892f748074b49 : SUCCESS in 8m 07s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/758e50418eb0416e805aedd8157761cf : SUCCESS in 4m 25s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb1a83590b3940878393dd15a6404358 : SUCCESS in 9m 21s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/524a29166d14453c913bb568b965070a : SUCCESS in 5m 21s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/752dc82cd9db44029558f2a883542858 : SUCCESS in 40m 41s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1fed3636cb04f2ca235ff3d0d573d4f : SUCCESS in 53m 05s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b58f11e3b0242fd8e7171e85d75f0cd : FAILURE in 1h 09m 38s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e31bef2ff884fedace6ff0ed25d5aeb : FAILURE in 51m 11s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/b083c73bff59412fbc0e43d471b003c5 : SUCCESS in 12m 24s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c069885d4c8a4066a2ec8daca0728787 : SUCCESS in 16m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24c6168899df4c97b9154a61859709e3 : SUCCESS in 16m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6301ecb9fbc648a885012259169f9e36 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d3c3b0aafb948b6fc3ea3500f9e42adbdd1dd385","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 15:57:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23026e15b261bbe8aec30ff669a9c5f2e1305b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 16:57:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b91baad4124af7855c2356ce1de85c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05370bbe252b47acaaaf43b2e88e4fb6 : SUCCESS in 1m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c12494ced054ee0a2a0bf5199ba50c9 : SUCCESS in 2m 12s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6003ea5142c04c90b652e0d66be0dc98 : SUCCESS in 4m 15s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c0f443158a644f382e5cf607a2dd3d0 : SUCCESS in 8m 52s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db01ebe6bca8443c897f63317830e195 : SUCCESS in 6m 55s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2701aa5785543d2939470b1d34388d8 : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aa1375169864f83b564a27b1b7533f5 : SUCCESS in 7m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63241b96d026459db7406fc253aba2ad : SUCCESS in 38m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a52b68b7d51e48c98aba80b76fbe8305 : SUCCESS in 51m 50s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/845824a4642847539537e41c92fb71f7 : SUCCESS in 39m 54s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/113c84244de94438af68b2983863ba60 : SUCCESS in 53m 49s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfe83ea82660407385d0b186886b6e57 : SUCCESS in 17m 46s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53dda7de1214473a96b844c1a3058d3a : SUCCESS in 12m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3b2b81de8cc4e538549922c5e474f93 : SUCCESS in 13m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82519e887f204c8ebe1a749552ee3af6 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":2},{"id":"934590484d605f61ea7b1d933df396bf77c45c8b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-28 07:14:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c2b7b46285acb644d1ef1c1aa97b87c652b0a4a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-28 14:12:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b686496ea5ca4bea11345f9c76b403d5f43ecd3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 14:13:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be2344f6eb81ab99e692a02ebea299caddb1a88d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 15:21:13.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14ec60a7b7e24a60bcffa74fcee9da77\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/275089dc97cc4959b9fbe4c37afb1769 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fbc3e9be9404c6882714bbae7f4ddba : SUCCESS in 2m 42s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c879142e0e6459592180f05bc7c06b5 : SUCCESS in 9m 25s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e58aba7d1f1b4ce48be670696d3e3fc4 : SUCCESS in 6m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/213c7bdcefd34a10b39542c2e77e850b : SUCCESS in 49m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49966602c590475982e91151413affd9 : FAILURE in 1h 01m 25s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/2faa08ee51824582a9b53b55cebf9fea : SUCCESS in 24m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1752cd058eb44e21bcd139caa8f369e9 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"13f9f72dbacb8a291040617521a66fb4beceaa47","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 01:10:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3c1e7bfdcbf9741560beef0d7435dbb4c44a0d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 02:14:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6af0f71dc91d4eb79f4427496af7b18c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94058c8978584647b0669ddb1ddede82 : SUCCESS in 1m 58s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcee8322dcf149d88cdba099e1104c5f : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe745e919f964beb91a960878dc4d14f : SUCCESS in 3m 32s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a5ffa12ca50424da16b196ac5a33c78 : SUCCESS in 8m 23s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8814f65b93945b1ae02a24bfc7eea67 : SUCCESS in 5m 28s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/94ac96b8e9cb4f5b89d38c1b61adcd55 : SUCCESS in 7m 03s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c64ecaae494a47148e3d74206bf8cc84 : SUCCESS in 5m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9508250df104452ea08a086f47abd758 : SUCCESS in 36m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61004f03cef046b9a2f0935056246497 : SUCCESS in 42m 52s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74c30f91a239402ab825ffcbaaf328bb : SUCCESS in 32m 37s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fbb98b78525489a8b060e73a6935fa7 : SUCCESS in 50m 15s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6565de17ae5d41ddbafbc925d4e5a04d : SUCCESS in 23m 07s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aa88e86d7f342e189cbabdfb298b314 : SUCCESS in 17m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc69b0e000a0489fb7f276eb360707d6 : SUCCESS in 16m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5109d0e6df44827be279a6407684135 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":2},{"id":"01a7978938c7d2ecba0c7429e5ea9377b2581a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 02:15:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bbfada764fd4a6c001bf714074c3653061af4e4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 03:23:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d907ef9ead1440e9a65710f5d7e72436\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea903886d11b4579a30d251b893cdc61 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5214f960ab04bd68049f26bd17c5c07 : SUCCESS in 2m 50s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cf646f38b654f7a82dfa0950a3568ce : SUCCESS in 10m 23s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7f4182c2583456db0a698f3403b5f8b : SUCCESS in 6m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d881ebbab38d41629b691ffb3db9cb85 : SUCCESS in 58m 04s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2acb411317334fcbbdeb258e0bf7e5c6 : FAILURE in 59m 23s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c2b250f0dd14e76bb7e3624075e115d : SUCCESS in 28m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f07bc9a54e04a13bac94d56bd8f7000 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"77c99e999a4f57e2dfbaa384a3654570f80a1d0a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 05:21:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1db8163abc696cad05069933df79587ade0c73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 06:26:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e8b2ebf904248a5b48a13d62a9359b9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc8cac99c0244d4d9222fe8390e21627 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbc8e1867ff44da59780bb1950fa1d08 : SUCCESS in 2m 16s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/567fb67385f34314bc046e9855b944f6 : SUCCESS in 4m 23s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ff4b4f7363a4b0dbb040687d42d0187 : SUCCESS in 7m 29s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b83c66eee6748e58b01312f9865a6db : SUCCESS in 6m 10s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ca8f3ceb4f243abbfe353cdd61802c3 : SUCCESS in 7m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f72773de358c4055b6b62ef12e831c74 : SUCCESS in 6m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94421f51b8394545bdb4f0e1ba5f6db0 : SUCCESS in 30m 52s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2562fe1ddddd43b18ad8aa9ba32e2f0a : SUCCESS in 41m 50s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56159b179a7946fbb22a15c42c09bff2 : SUCCESS in 30m 59s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1668d284ffa442d8419d8cb6ef4e472 : SUCCESS in 33m 22s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c88264c134e421c8be60eb1e766bbe9 : SUCCESS in 19m 25s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d48ad0ba0302446cbfbcc9344d852e56 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a3adf0ec2cb42c0a36d969661cc6c9a : SUCCESS in 10m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edb43191302a46f0abde1051f777753a : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"187edd378848cd620eddef7eb437cc6a02c626e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 06:27:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6bd644c4c6afc6fb9f65843ffe4d227a9823ddca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 07:26:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f9425e7c3d341f29c74352b53ba686f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f417ff7c2ef4533b97c069c104cf61d : SUCCESS in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11bbc09ab54649928270e48b0a577a0e : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/276a677fe188481ead757fa8b1b6ba23 : SUCCESS in 5m 40s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3919e2702d5040e295954c2d12ff30fd : SUCCESS in 4m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ae2a9da5be34c05aeb0e3d3f65e3987 : SUCCESS in 52m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f62efa92c86640bf9e405b2f0bee5b6f : SUCCESS in 57m 30s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/a33c6287f6584f8da79354715479d282 : SUCCESS in 24m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5fe1a16f5fb48f9a8d7507efb803aa8 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"9f2599cda3bca8b0df0c32ba085c70e36c22f39d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 07:26:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1f77021b0c9e9a102a926f6e2e78440404841e28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 07:27:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcf83f251f634602af9a182dadbce297\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec2c31d1d19f4fb39cb7d3f7a1c2e80a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"51b49ea18f698ee83f8acde4913998c82920559c","revisions":{"65daa600703203212224f43388ce62beb657e2de":{"kind":"REWORK","_number":1,"created":"2024-10-26 12:53:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/933490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/933490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/1"}}},"commit":{"parents":[{"commit":"57f3352a64da436333f65403172d0cc5ae7ce811","subject":"Expose configure_service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/57f3352a64da436333f65403172d0cc5ae7ce811"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 12:52:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 12:52:57.000000000","tz":540},"subject":"Add support for [task_flow] jobboard_redis_backend_db","message":"Add support for [task_flow] jobboard_redis_backend_db\n\n... which was added during the Epoxy cycle.\n\nDepends-on: https://review.opendev.org/923702\nChange-Id: Ib7bcf34cb74135f5ff5f9ad8ba77d707e493f24a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/65daa600703203212224f43388ce62beb657e2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/65daa600703203212224f43388ce62beb657e2de"}]},"branch":"refs/heads/master"},"51b49ea18f698ee83f8acde4913998c82920559c":{"kind":"REWORK","_number":2,"created":"2024-10-26 13:25:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/933490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/933490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/933490/2"}}},"commit":{"parents":[{"commit":"57f3352a64da436333f65403172d0cc5ae7ce811","subject":"Expose configure_service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/57f3352a64da436333f65403172d0cc5ae7ce811"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 12:52:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 13:25:40.000000000","tz":540},"subject":"Add support for [task_flow] jobboard_redis_backend_db","message":"Add support for [task_flow] jobboard_redis_backend_db\n\n... which was added during the Epoxy cycle.\n\nDepends-on: https://review.opendev.org/923702\nChange-Id: Ib7bcf34cb74135f5ff5f9ad8ba77d707e493f24a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/51b49ea18f698ee83f8acde4913998c82920559c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/51b49ea18f698ee83f8acde4913998c82920559c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
