)]}'
{"id":"openstack%2Fpuppet-octavia~941596","triplet_id":"openstack%2Fpuppet-octavia~master~Ic83ce7512077e0c5247222ec8214364954914b9b","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-14 17:11:45.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-02-14 17:11:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic83ce7512077e0c5247222ec8214364954914b9b","subject":"Make password in octavia::neutron class required","status":"MERGED","created":"2025-02-13 20:05:41.000000000","updated":"2025-02-14 17:12:42.000000000","submitted":"2025-02-14 17:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941596","meta_rev_id":"f5153c06d80fda32cfaf666c56a950ddbc25879d","_number":941596,"virtual_id_number":941596,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-14 17:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-02-14 12:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-02-14 15:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-02-14 15:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 22:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 12:39:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-02-14 15:32:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"91c4fecf70bc416235ba927bbc2bba77e56b94f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-13 20:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a27e87ae0ba4086162344d5d209723102e237890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 22:13:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ab56b3d64304478be23c34f9c239390\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1006527b7104a718a3a3970121e1267 : SUCCESS in 2m 10s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3ad65d0d4f44966af3d5113368f845a : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d200ad5a3faa4302a9f27cb0d5172431 : SUCCESS in 3m 27s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2d196adf4374c9ebdd9036f6511d094 : FAILURE in 11m 00s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a012ab29c74f558c1139c669c86089 : FAILURE in 7m 11s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/aff85efc64e74056a9a4dbac3d65c1cd : FAILURE in 9m 03s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55ae47527b6a482ebc901096d33d4492 : FAILURE in 6m 13s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c27b55950b1743f782aecf28d2806204 : SUCCESS in 27m 45s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/754aa8f572c347b0ba27f43cb8897302 : SUCCESS in 1h 05m 06s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/375dd1891e0d4123a47563326d7b4cd1 : SUCCESS in 1h 04m 35s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6eb9b493520141278ddcca83b15a88f8 : SUCCESS in 47m 32s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5726de6a69c24d07b809cb0993637aa9 : SUCCESS in 42m 15s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71b08a32335f414abb01a57d5ee59424 : FAILURE in 55m 24s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0963da6a1d0849b380f7edce292c440f : SUCCESS in 41m 41s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/328bbf09f21244d2addb25f84c701fe6 : SUCCESS in 17m 22s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbdf47eb48854d3eace49e1ba9fbc27e : TIMED_OUT in 1h 31m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64243c310ed04cd8901c3fe3c51df440 : SUCCESS in 15m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c714090de3aa4e5194addc153945cdec : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fbe1a9f02b9ef5342d1cafd357e0655dc023bc86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-14 00:11:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d53ce4f08d4a2a16726b463bcf3d53eef033d070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 01:43:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30968fd5af3443b597055307bab96f1d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbe08b76b4c149078c99568b3139dd78 : SUCCESS in 2m 16s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2b119e8463b4c3d83955bb6622db3b9 : SUCCESS in 1m 57s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/af345a1e8edc4fa3a9e4635497b7cc1c : SUCCESS in 4m 46s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/23c4432c9bfd4098ade697da7ad9fffb : FAILURE in 9m 51s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f46e4cf4339d414b99b53b6fe5ef1008 : FAILURE in 4m 23s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3786dd296bf845b5807e71d72a61134b : FAILURE in 7m 40s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aecb989a99d946b383d2ffe3f684041f : FAILURE in 4m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/977b3169d14f48abbb37d326d1c163af : SUCCESS in 37m 56s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/699acf1d89ed4af695020d81289acb3c : SUCCESS in 1h 01m 04s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fa8e304427b454c80720335229e8721 : SUCCESS in 1h 00m 10s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/672c8d8876d74ffe93cb11c6c34b2322 : SUCCESS in 24m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/09f312db05aa4c2c97536f75a2bc6e54 : SUCCESS in 1h 18m 08s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/750bfd991e754daf962895b1eb3f5d52 : SUCCESS in 1h 22m 52s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b20758272334af18483310e573528c7 : SUCCESS in 37m 09s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1692bf329f04daeb577743933809ac3 : SUCCESS in 17m 24s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/42bfda362ea1441ba0904e06f1a303e6 : TIMED_OUT in 1h 31m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a504419862c649f1b437bf66315fd5a1 : SUCCESS in 16m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7688c4923dee4e6d96426b8a3eb93631 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":2},{"id":"14a9e74e6a05dd3018962f47389b318c62e2772d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-14 09:08:58.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a189f71aede4f05d5310b462a814a62cf280dcc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-14 09:09:51.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3d9a5324363d5d903f0eee053ce59d4ce97bc9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 10:46:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b443720343148cb8db28a603291b3f9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7220f72eb4344e76b9f9e4e86a6ab4fa : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16afce53397e4a77a91efc4ece72292a : SUCCESS in 2m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6dd20eae978649a2b83e785dcec440e4 : SUCCESS in 5m 58s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8884c965b0a4b78a259e534793f831f : SUCCESS in 11m 33s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6684f3401f94453eb370027afc796995 : SUCCESS in 5m 43s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a533f054e9c40e697a322690862c935 : SUCCESS in 10m 00s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8add51cfd8fd4039b663a78730f10429 : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48036316b9874b99804ab36bab76d98b : SUCCESS in 41m 04s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/359f62cef3a9420385424c00ad9dcd1e : SUCCESS in 1h 05m 18s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0fd820277c6470ab450d754d45a9ad9 : SUCCESS in 1h 04m 56s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d4d8ea2a15a46bb9682415a69e8def9 : SUCCESS in 41m 00s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a0754ddb7024322ac0dbcb10f91e3a2 : SUCCESS in 1h 15m 47s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d8df8d2c9c945c8af6c9e78fa6de774 : SUCCESS in 1h 04m 28s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/06446c4488684edfb7ba3f8d61470aaf : SUCCESS in 42m 19s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e87ec35038864c42b1a1abf713d1c0ac : SUCCESS in 16m 04s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c30d5b1e5e2043b3ad4f9dff1cd94be1 : FAILURE in 38m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2694a3024584d4aad2a9bea95943c58 : SUCCESS in 17m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83951eb69fcc449983a377c4f5bed780 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":4},{"id":"4c6b63d3d89e7b18295a246b6db82d04e9fd96dd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-14 12:39:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4e42909b61831453ef4c2c6266a302effc3b3188","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-14 15:32:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"25eccdc4d4e75a4f6136ee63556cbf4ed129630b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 15:33:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b17f8669092dc8fba1d33abb48ae46c2ad975582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:11:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840d3fc9d5b54166a7318cdba2738911\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b70f641073f4022945ad78c604d1ca5 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4b6f0a055714ba4a60409bfc19d61d2 : SUCCESS in 3m 16s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/758d50f7edbb4c9b89f15e7ef9fcaf7c : SUCCESS in 9m 59s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1c69fb7e328496f9a98c8a6ddc7a231 : SUCCESS in 7m 31s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/91ed67cf05174e82b55f804ee9ef633b : SUCCESS in 11m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb5f6727d8c64247a7c21645c6910826 : SUCCESS in 1h 00m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80c1112e5f5241a4a797945df8aa658a : SUCCESS in 1h 22m 21s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/205975f36c6e497aa596f6ec64f7077a : SUCCESS in 45m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5686650dfa204b53bb8c7a84a446357a : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d38cdacade1c8918a43b344fb40627d3d6fb7e98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:11:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f5153c06d80fda32cfaf666c56a950ddbc25879d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:12:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f692c9814c84e709552d667c0d36f36\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fdfe2240f0c47f9a0d634c571f190d2 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e5e9a77646c7d7fd71ba494ea9674bd783c44c7d","revisions":{"795d4c1ff804d7b7c84e3baaec109478a10c0cbd":{"kind":"REWORK","_number":1,"created":"2025-02-13 20:05:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/96/941596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/96/941596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/1"}}},"commit":{"parents":[{"commit":"4598e037119fb5a01389430bac7a8ffe6cf23cf6","subject":"task_flow: Hard-code default jobboard_backend_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4598e037119fb5a01389430bac7a8ffe6cf23cf6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-13 20:04:39.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-13 20:05:39.000000000","tz":60},"subject":"Make password in octavia::neutron class required","message":"Make password in octavia::neutron class required\n\nThis makes the password parameter required\nin the octavia::neutron class and also\nmentions that its a required class for\nOctavia to work correctly.\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/941595\nChange-Id: Ic83ce7512077e0c5247222ec8214364954914b9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/795d4c1ff804d7b7c84e3baaec109478a10c0cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/795d4c1ff804d7b7c84e3baaec109478a10c0cbd"}]},"branch":"refs/heads/master"},"513eedd1fbef68b2227769154310fb190d06a359":{"kind":"REWORK","_number":2,"created":"2025-02-14 00:11:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/941596/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/96/941596/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/2"}}},"commit":{"parents":[{"commit":"4598e037119fb5a01389430bac7a8ffe6cf23cf6","subject":"task_flow: Hard-code default jobboard_backend_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4598e037119fb5a01389430bac7a8ffe6cf23cf6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-13 20:04:39.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-14 00:11:38.000000000","tz":540},"subject":"Make password in octavia::neutron class required","message":"Make password in octavia::neutron class required\n\nOctavia no longer uses service_auth section to load credential used to\ninteract with neutron, and it requires credential in [neutron] section\nsince 2023.2[1].\n\nThis makes the password parameter required in the octavia::neutron\nclass and also mentions that its a required class for Octavia to work\ncorrectly.\n\n[1] 3e6fd13bfb000785d5eb9efe2139ee739e074460\n\nDepends-On: https://review.opendev.org/941595\nChange-Id: Ic83ce7512077e0c5247222ec8214364954914b9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/513eedd1fbef68b2227769154310fb190d06a359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/513eedd1fbef68b2227769154310fb190d06a359"}]},"branch":"refs/heads/master"},"ae1581d3ff0271e7b1d3dac84d28d52552f72b1f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-14 09:08:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/941596/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/96/941596/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/3"}}},"commit":{"parents":[{"commit":"4598e037119fb5a01389430bac7a8ffe6cf23cf6","subject":"task_flow: Hard-code default jobboard_backend_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4598e037119fb5a01389430bac7a8ffe6cf23cf6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-13 20:04:39.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-14 09:08:58.000000000","tz":0},"subject":"Make password in octavia::neutron class required","message":"Make password in octavia::neutron class required\n\nThis makes the password parameter required in the octavia::neutron\nclass and also mentions that its a required class for Octavia to work\ncorrectly.\n\nDepends-On: https://review.opendev.org/941595\nChange-Id: Ic83ce7512077e0c5247222ec8214364954914b9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ae1581d3ff0271e7b1d3dac84d28d52552f72b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ae1581d3ff0271e7b1d3dac84d28d52552f72b1f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e5e9a77646c7d7fd71ba494ea9674bd783c44c7d":{"kind":"REWORK","_number":4,"created":"2025-02-14 09:09:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/941596/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/96/941596/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/96/941596/4"}}},"commit":{"parents":[{"commit":"4598e037119fb5a01389430bac7a8ffe6cf23cf6","subject":"task_flow: Hard-code default jobboard_backend_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4598e037119fb5a01389430bac7a8ffe6cf23cf6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-13 20:04:39.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-14 09:09:51.000000000","tz":0},"subject":"Make password in octavia::neutron class required","message":"Make password in octavia::neutron class required\n\nThis makes the password parameter required in the octavia::neutron\nclass and also mentions that its a required class for Octavia to work\ncorrectly.\n\nDepends-On: https://review.opendev.org/941595\nChange-Id: Ic83ce7512077e0c5247222ec8214364954914b9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5e9a77646c7d7fd71ba494ea9674bd783c44c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5e9a77646c7d7fd71ba494ea9674bd783c44c7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
