)]}'
{"id":"openstack%2Fpuppet-octavia~941658","triplet_id":"openstack%2Fpuppet-octavia~master~I85bfa4944c979c97e711e5c8ef7765b259ccc28b","project":"openstack/puppet-octavia","branch":"master","topic":"puppet8","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-14 17:11:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85bfa4944c979c97e711e5c8ef7765b259ccc28b","subject":"Mark puppet 8 supported","status":"MERGED","created":"2025-02-14 13:51:46.000000000","updated":"2025-02-14 17:13:31.000000000","submitted":"2025-02-14 17:11:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941658-puppet8","meta_rev_id":"5a2a3b18fba5ab9f82799ed3a72c9856856aeb65","_number":941658,"virtual_id_number":941658,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-14 17:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-02-14 14:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-02-14 15:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-02-14 15:32:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-14 13:57:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 14:38:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-02-14 15:32:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4c0de5c05fb9538e58704e300bce860409c1e183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-14 13:51:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15a6ed199bcba0de0701c71e4ba2ddaf13a8c1eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 13:57:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46a922d5a9b841d6a5087903ebfcf391\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/e4e5225bef36423f9bb85cd9611a5a67 : SUCCESS in 2m 41s\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/372f2a77b22c4f24af414434e608ea9a : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc0f4c21479a496fafae2e39ffc14a02 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d09264119575de0831c8d32c4aed01977df923e7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-14 14:38:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bac2771d5c268d2e30a39ba0f9c741ce6027cfaa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-14 15:32:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9176a73c7fa91dc145ec67cef761ee51bfe8abf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 15:33:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6819d51fc78f357c9ab5a96dfede13ef172f37d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:11:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d2e8f5c92f949d5adc5f74e44b863f4\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/84fad91aec1f4da1a5826c60357219f0 : SUCCESS in 2m 15s\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0de83d4d7dd34ede93a15187ae6651b4 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3dbc3a61aa49c8888c0c98fef34f59 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4c9ce0ef88acb98ff524402b0049be2b073d7669","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:11:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a2a3b18fba5ab9f82799ed3a72c9856856aeb65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:13:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71fe2aa9815e4a3094348f5a6ddcc159\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d05e7da94d0e4e7f81b504e0fba8a6ff : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bf714ad0cc0c51f6095c6f03c8fdc4437bac9ea","revisions":{"5bf714ad0cc0c51f6095c6f03c8fdc4437bac9ea":{"kind":"REWORK","_number":1,"created":"2025-02-14 13:51:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/941658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/58/941658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/58/941658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/58/941658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/58/941658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/58/941658/1"}}},"commit":{"parents":[{"commit":"4598e037119fb5a01389430bac7a8ffe6cf23cf6","subject":"task_flow: Hard-code default jobboard_backend_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4598e037119fb5a01389430bac7a8ffe6cf23cf6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-14 13:51:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-14 13:51:35.000000000","tz":540},"subject":"Mark puppet 8 supported","message":"Mark puppet 8 supported\n\nNow we test puppet 8 in CI so can declare that we support it.\n\nChange-Id: I85bfa4944c979c97e711e5c8ef7765b259ccc28b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5bf714ad0cc0c51f6095c6f03c8fdc4437bac9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5bf714ad0cc0c51f6095c6f03c8fdc4437bac9ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
