)]}'
{"id":"openstack%2Fpuppet-octavia~942042","triplet_id":"openstack%2Fpuppet-octavia~master~Ib4c899a079ca706738c2bfd9aa18fb6d13d9d26a","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-19 16:11:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4c899a079ca706738c2bfd9aa18fb6d13d9d26a","subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","status":"MERGED","created":"2025-02-18 02:01:47.000000000","updated":"2025-02-19 16:12:43.000000000","submitted":"2025-02-19 16:11:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942042","meta_rev_id":"cac3a6d50b849537545c77f036e446aae626dd02","_number":942042,"virtual_id_number":942042,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-19 16:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-02-18 15:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-02-19 14:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-02-19 14:16:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-18 04:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 15:08:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-02-19 14:16:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4e4327195f8c143f42db246dcc48bf819c0ae9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:01:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cedf1a2564ca8e1a1947a05ed28164ebb242c1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:34:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d586b2d16282e69d1d309a090dd10e8c6c62ca6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:34:42.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f450244a0b59d0701e2d802da599ab638364ebc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:57:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"99f394c0290f493406c7fff122fd5f77341b2869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 04:43:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b2ba195edc4f20ba74b6843ed17436\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3adc1eeee79c43b9ac78ac33bc434aec : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d58437b012742cfac02e8539127ed34 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9776abe98f894224a5de1174745759ac : SUCCESS in 4m 42s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9a81419d711411e9f7bd136b94bfd32 : SUCCESS in 12m 44s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df0b485e31e6410fa029ad09ffefbdf2 : SUCCESS in 6m 12s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5404138e9a0e48b6a5fccc8032a01207 : SUCCESS in 9m 09s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/270556ef965c4fa3b3de3fa542ac35d1 : SUCCESS in 5m 18s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fa2ba5887a94b3ea44c09a26a2b7245 : FAILURE in 1h 07m 08s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b93d22325a44a109ad62f2f817bc71c : SUCCESS in 56m 38s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7cad5cade7745ae85c9d5d30fba1c3e : SUCCESS in 57m 54s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5c56d9ef6e545b39981fc1b16583d30 : FAILURE in 26m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2716c62867454369ba0c208a33f3879f : SUCCESS in 1h 22m 30s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0c1ad3855e44b95a0d89230d7519637 : SUCCESS in 1h 01m 25s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e7124518e5344c484a10473e66a9255 : SUCCESS in 49m 00s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96d1ee02ebda48628200efe2e4ec3637 : SUCCESS in 17m 33s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b00f86d7dae14745b7351217b67516ee : SUCCESS in 46m 55s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/307620affaf3423180a2a6bc5e3abd2b : SUCCESS in 16m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c76c931b3084388899be1b2e87a3d7f : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":4},{"id":"81e275860049778472ee078b28c9b7b13c6b7753","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-18 15:08:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"970ef6a95e855a6e4cec2a72c2d442561e7d0bcc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-19 14:16:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"31644c0f84636823b770b4e41eceef13aa5822ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 14:17:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"635c0c968c21196b26db8169d457ec91a44dea45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:11:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd7c946396584439b496636052f385bb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/369afd095aa540d4bf1801682e85f7b6 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d7104f0cbdb4888a52c64e8bfd67a88 : SUCCESS in 2m 23s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/acba62e8c03841599f44d944f6d51e74 : SUCCESS in 8m 22s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b13aaef0e9754f6d85eb435fbfbd1c72 : SUCCESS in 6m 19s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4f28364b32b45a480199b57e6ff5ef5 : SUCCESS in 9m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e3c1e4a4f704078bef962dbadd7362a : SUCCESS in 1h 25m 49s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9175ae2ae014944849c6bc9ffa0ccf7 : SUCCESS in 1h 36m 57s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/8641234731964e96b6345ca0a6bc8d5b : SUCCESS in 54m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdfdd9679c144f30a2272942dd730656 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"2bfbcaf62b21480c1b7223fb88627773e0fea99d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:11:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cac3a6d50b849537545c77f036e446aae626dd02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:12:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c049ac8022164ef4b3cc8d882e3dfafb\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab8fef7aa42486fb025e4c2dd48fb8c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"52928b4c9d5bebd8aaf1a2c5b5997e4852e22d31","revisions":{"06cca3630125362a3ff87c0b7bccad6b7b959932":{"kind":"REWORK","_number":1,"created":"2025-02-18 02:01:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/942042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/42/942042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/1"}}},"commit":{"parents":[{"commit":"cd1f4862fd2421984a8b70ee6a1151811e062ad4","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cd1f4862fd2421984a8b70ee6a1151811e062ad4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:01:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:01:35.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ib4c899a079ca706738c2bfd9aa18fb6d13d9d26a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/06cca3630125362a3ff87c0b7bccad6b7b959932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/06cca3630125362a3ff87c0b7bccad6b7b959932"}]},"branch":"refs/heads/master"},"2dbc7994fe90885e950559bc31fedba53b875da2":{"kind":"REWORK","_number":2,"created":"2025-02-18 02:34:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/942042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/42/942042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/2"}}},"commit":{"parents":[{"commit":"cd1f4862fd2421984a8b70ee6a1151811e062ad4","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cd1f4862fd2421984a8b70ee6a1151811e062ad4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:01:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:34:20.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ib4c899a079ca706738c2bfd9aa18fb6d13d9d26a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2dbc7994fe90885e950559bc31fedba53b875da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2dbc7994fe90885e950559bc31fedba53b875da2"}]},"branch":"refs/heads/master"},"1eb401aab66b6c4b3d4435e04438aa61607c3556":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-18 02:34:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/942042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/42/942042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/3"}}},"commit":{"parents":[{"commit":"cd1f4862fd2421984a8b70ee6a1151811e062ad4","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cd1f4862fd2421984a8b70ee6a1151811e062ad4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:01:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:34:32.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ib4c899a079ca706738c2bfd9aa18fb6d13d9d26a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1eb401aab66b6c4b3d4435e04438aa61607c3556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1eb401aab66b6c4b3d4435e04438aa61607c3556"}]},"branch":"refs/heads/master"},"52928b4c9d5bebd8aaf1a2c5b5997e4852e22d31":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-02-18 02:57:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/942042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/42/942042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/42/942042/4"}}},"commit":{"parents":[{"commit":"cd1f4862fd2421984a8b70ee6a1151811e062ad4","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cd1f4862fd2421984a8b70ee6a1151811e062ad4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:01:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:57:16.000000000","tz":540},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ib4c899a079ca706738c2bfd9aa18fb6d13d9d26a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/52928b4c9d5bebd8aaf1a2c5b5997e4852e22d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/52928b4c9d5bebd8aaf1a2c5b5997e4852e22d31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
