)]}'
{"id":"openstack%2Fpuppet-octavia~944213","triplet_id":"openstack%2Fpuppet-octavia~master~I676917dce07b6490cb34eba876be48a67194dea0","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-14 15:44:43.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-03-14 15:44:43.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-03-14 15:44:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I676917dce07b6490cb34eba876be48a67194dea0","subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","status":"MERGED","created":"2025-03-13 11:45:31.000000000","updated":"2025-03-14 15:45:46.000000000","submitted":"2025-03-14 15:44:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944213","meta_rev_id":"5510116b0d9e30c45f70255da6f33756d21aa33b","_number":944213,"virtual_id_number":944213,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-14 15:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-03-14 04:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-03-14 14:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-03-14 14:42:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-13 12:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 14:11:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-03-14 14:28:50.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-03-14 14:42:27.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a49dd25c8e24be5087011ba344cf9db9462f8797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-13 11:45:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20161972b2a8ed43b135d9b8b857033244075322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 12:37:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f06d9ce737404675b65f0cee4fc699ae\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/859396e216da4e838b9209ac9fe8b832 : SUCCESS in 2m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e9ccc8c763b43de875f48532a0770fe : SUCCESS in 1m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0d846c659a44f0894c9ff2e2b076aa1 : SUCCESS in 3m 05s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc0d7195dd184fa681a47abd6d535507 : SUCCESS in 4m 25s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7720f99510bd4a4ebd6af99e9f6986d0 : SUCCESS in 3m 08s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d65d2a9a7374cb1a65f5a149ea86698 : SUCCESS in 4m 37s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f54d3ed026b74c35970a8ba73a96bf9a : SUCCESS in 3m 10s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9481bfa5f09c48f18601fa8f847cc973 : SUCCESS in 18m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/908c51df0e1b43018fdcc33a53e6fc77 : SUCCESS in 29m 08s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a743516761f4eaab3f99859eb46cf04 : SUCCESS in 29m 38s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d47c8eef53f9461a8917db73f6ecf997 : SUCCESS in 24m 52s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6acee0201d0f4dd98fdba06a33a90736 : SUCCESS in 31m 03s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc786224404e462fb51f06a3d1166bdc : SUCCESS in 49m 58s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9ca16602e42454e84db9a8b6a2767b7 : SUCCESS in 24m 27s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6746912b4224ab28e874c3da655b0c3 : SUCCESS in 9m 34s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e3a20f408314ea088d0be1751e03867 : SUCCESS in 25m 45s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e91b2ba871fe458c8c77d1a6054e33d8 : SUCCESS in 16m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bd519e759064444a7447796d92f554f : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"67543d20377322578aa76baace653ca72f3719aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-13 14:11:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ea661c4b763ca3a3778d81dd56b35682a3f14668","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-13 15:31:04.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"df4608bcb48f8d4ead194534f262ac67bddfdaf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-13 15:31:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a219c40b6c4dc1b03f876135fbf2b1f02d430fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 16:31:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac15194c8b7c4d3e82f9826503d42eff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7848999afaa842f2832e951c527505ad : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2283ebbfc94873bb0a38ef86377ee8 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3692552159374f5b9a35c257ada85e01 : SUCCESS in 5m 36s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c8c4a468e244de5937be99b2274dcc1 : SUCCESS in 9m 37s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f58ab437c8524ee981e8847e26826429 : SUCCESS in 6m 38s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2acaaabd8f21413793e4284c7562743d : SUCCESS in 9m 21s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15f1878f4cdd49c4905eeda1a8d0856d : SUCCESS in 6m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ffc15aa1d4647ec98ecf983bbe30e0e : SUCCESS in 40m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/83fbf31218b84ac0844c1d2dc1224ca1 : SUCCESS in 49m 27s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb4e0818b47d4692a4a528cbd3d06c7b : SUCCESS in 50m 58s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adca8909976b4b42b5b4a726cfba9a84 : SUCCESS in 43m 47s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79c6cfbb47284a95b4e1116ad2bf1a72 : SUCCESS in 55m 12s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b719ddc1119457487e99c9d32be1fde : SUCCESS in 56m 03s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ac9ece3e2a7449385e972830a8866e1 : SUCCESS in 24m 24s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78b1c5645bbc496ca2586c724277d079 : SUCCESS in 15m 26s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/632ce18832794650be6cc0f6a307677a : SUCCESS in 23m 13s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36e86b5ae6114762bb5903c024ef7330 : SUCCESS in 18m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd3aae81a25446bbbd32d4c170bfd917 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a17ce4af6679e217057f76c576a5add5443ef67d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-14 00:32:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4140b1d69d996620b20a83fe9894483792cd76b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-14 04:08:23.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"434590a88f5d140eb8c4b7a2f9f94e4b3826106b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-14 04:08:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a0ae45c944319221a2d534fb28f2ba370dc6bda6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 05:19:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10e6d8a71f0d45a8adb3d4265f45dd5c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/366ce69e4ca0436789faf3eb599f5c87 : SUCCESS in 1m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d9556aa65f3429dabd4a48893c69f61 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ccb33773fc249669d08ad77488d51ff : SUCCESS in 5m 07s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f229a1b062e4cd6ae987498c0bc61fc : SUCCESS in 11m 12s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acdf5814f8344b5fbe502816808fe3f4 : SUCCESS in 5m 44s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/915a97fe0cd641da8c7d26c8d06e7751 : SUCCESS in 9m 12s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45a199bde814415180beb526f8ed2135 : SUCCESS in 6m 58s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be2082a244e44366b2da60851c7bd1db : SUCCESS in 38m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b7dc1d451d241ea83045fcda07aba0a : SUCCESS in 57m 07s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e9b5c374c6f4e6fa1431e3d0ff8d314 : SUCCESS in 46m 03s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a90013932b041b58a9d8817af4ed7a9 : FAILURE in 26m 07s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/381227faa6cb4ceeafd9388c89a3012d : SUCCESS in 52m 47s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f3c1de0bd904185aae2a4fdc7822ced : SUCCESS in 51m 02s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d052517399af466f9bd2d1b18a0eac20 : SUCCESS in 23m 54s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/260101dca2ce4b939342a37af26d196a : SUCCESS in 16m 53s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cede0bf6aab479b85aeec3d73a098e3 : SUCCESS in 26m 29s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97c1fd2e58264787b0edacc459c355e0 : SUCCESS in 17m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9592f9ad4944fea9a8f47f86b3a57ac : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3c53836803fdd7f2e117acb2a174e6cac265b8dc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-14 14:42:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e8ea0b969aec8c3ed5275fa204fa5e6efe46a6da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 14:43:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2b510aca4bfefa12864ca88e0d286fc223e63484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:44:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/560a7a85b3ce4054bcbfde3f03fa0871\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d493904626943e08ba7c192a1866df3 : SUCCESS in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b50f40bc8dc44e29a940980e6a642e9 : SUCCESS in 2m 27s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/a829930a2ef6455d8a453e499aabcd5e : SUCCESS in 10m 10s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75d54c521b454cad8e47832d308aacf7 : SUCCESS in 6m 26s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6386c9e8fd4443a480a238c4419cbd96 : SUCCESS in 5m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b461699d619b46c494ed7f7ad5677d15 : SUCCESS in 56m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d62bec8765ef493c817d55fbb30b0c6e : SUCCESS in 55m 26s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/51358e1f959f49b5a0ddf25cfcc26739 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dab6b270ff6e4e6a82b19e56c37ec0b8 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"7958824ee12ec50d236a6465c03e50bd80c85f31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:44:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5510116b0d9e30c45f70255da6f33756d21aa33b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:45:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a86bbbb85940aea7c4c29f18c8cc5d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ee0a9c1101f4892b7e9ee1dc31ecb75 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4504420e4f45ef64097425f918cc609915c6ba66","revisions":{"287ef700764e0ef76bd1177a9c1d884d0dd51395":{"kind":"REWORK","_number":1,"created":"2025-03-13 11:45:31.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/944213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/13/944213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/1"}}},"commit":{"parents":[{"commit":"f1363fc8df6e9749d42de24616aed8a1b6d6fafd","subject":"Fix new lint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f1363fc8df6e9749d42de24616aed8a1b6d6fafd"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-13 11:44:46.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-13 11:44:46.000000000","tz":60},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nThe amqp_auto_delete option needs this patch:\nhttps://review.opendev.org/c/openstack/puppet-oslo/+/944132\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use octavia_config to set it, so this patch is\nmandatory for one to set rabbit_transient_queues_ttl in octavia.conf.\n\nChange-Id: I676917dce07b6490cb34eba876be48a67194dea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/287ef700764e0ef76bd1177a9c1d884d0dd51395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/287ef700764e0ef76bd1177a9c1d884d0dd51395"}]},"branch":"refs/heads/master"},"bbb0e4f70356bede63d4071bf4eb91bd4ce4aa7d":{"kind":"REWORK","_number":2,"created":"2025-03-13 15:31:10.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/944213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/13/944213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/2"}}},"commit":{"parents":[{"commit":"f1363fc8df6e9749d42de24616aed8a1b6d6fafd","subject":"Fix new lint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f1363fc8df6e9749d42de24616aed8a1b6d6fafd"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-13 11:44:46.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-13 15:30:57.000000000","tz":60},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nThe amqp_auto_delete option needs this patch:\nhttps://review.opendev.org/c/openstack/puppet-oslo/+/944132\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use octavia_config to set it, so this patch is\nmandatory for one to set rabbit_transient_queues_ttl in octavia.conf.\n\nChange-Id: I676917dce07b6490cb34eba876be48a67194dea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bbb0e4f70356bede63d4071bf4eb91bd4ce4aa7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bbb0e4f70356bede63d4071bf4eb91bd4ce4aa7d"}]},"branch":"refs/heads/master"},"4504420e4f45ef64097425f918cc609915c6ba66":{"kind":"REWORK","_number":3,"created":"2025-03-14 04:08:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/944213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/13/944213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/13/944213/3"}}},"commit":{"parents":[{"commit":"f1363fc8df6e9749d42de24616aed8a1b6d6fafd","subject":"Fix new lint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f1363fc8df6e9749d42de24616aed8a1b6d6fafd"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-13 11:44:46.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 04:08:22.000000000","tz":0},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nThe amqp_auto_delete option needs this patch:\nhttps://review.opendev.org/c/openstack/puppet-oslo/+/944132\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use octavia_config to set it, so this patch is\nmandatory for one to set rabbit_transient_queues_ttl in octavia.conf.\n\nChange-Id: I676917dce07b6490cb34eba876be48a67194dea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4504420e4f45ef64097425f918cc609915c6ba66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4504420e4f45ef64097425f918cc609915c6ba66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
