)]}'
{"id":"openstack%2Fpuppet-octavia~953024","triplet_id":"openstack%2Fpuppet-octavia~master~I75037abd5f0d17eb50d97b8355509c782325c788","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-06-23 21:20:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I75037abd5f0d17eb50d97b8355509c782325c788","subject":"Add queue_manager and stream_fanout","status":"MERGED","created":"2025-06-20 21:54:55.000000000","updated":"2025-06-23 21:21:06.000000000","submitted":"2025-06-23 21:20:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953024","meta_rev_id":"13ae089cb6e116f645bcf23b70171426d71632f5","_number":953024,"virtual_id_number":953024,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-23 21:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"all":[{"value":1,"date":"2025-06-23 06:20:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":2,"date":"2025-06-23 06:54:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-06-23 20:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-06-23 20:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-20 22:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 06:20:37.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"REVIEWER"},{"updated":"2025-06-23 06:54:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-06-23 20:45:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4316bd40200989c1176cd73c62add9acb2a58a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-20 21:54:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74fb5767a8ed775704ec0f3451db4ed79ce0370a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 22:59:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/840dff00b3dc468e96f3ddadc82fb811\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54296dfee36e4aa2a75e4279f0a17a1e : SUCCESS in 2m 10s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56b341f7a03f47d9af5cc68acc1a044e : SUCCESS in 2m 15s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18ae182752844aa4880df438a7beeb10 : SUCCESS in 3m 30s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a77dae2de064bfa977f443ce664d6ef : FAILURE in 8m 56s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bac1dded981a41598d06666079aff233 : FAILURE in 5m 52s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4239342ccc0f44d383c57a9c838fab2e : FAILURE in 5m 30s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3884191156e4daebb40780bec6965d8 : SUCCESS in 35m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/83b392fbf8e64e079c833a682207bd6e : SUCCESS in 51m 02s\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f081ca5f31174195a8b32c0b6315aad1 : SUCCESS in 34m 03s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2d8f51e1e2a4793b3b2e6f67dde1f10 : SUCCESS in 53m 36s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c3b2ca491d548c98edf633f386a9473 : SUCCESS in 23m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da887c799d504093a7436cf5e43e084a : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26886d8fea964c6bbfa4b954772a09a8 : SUCCESS in 16m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c06254b65fe4078a5a2df640b5e493a : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2d62461321f30e4335be1b9865da8cdd0a47db38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-20 23:07:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d1a979df00e5e704d5a38348bd2f7ee701e61a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 23:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ea9e06eb62435b98be8e2fa7c47ce0\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e8c110400344bc2a92e93cc974f8b27 : SUCCESS in 2m 04s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69cf001ef1a24c9fbb21ede2da9e64dd : SUCCESS in 3m 07s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/475d1ae4b8a14a588e189ef7f1c91c31 : SUCCESS in 2m 07s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/891953b2e71b4ab0bbd93598cf659e8a : SUCCESS in 6m 32s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38ae70f00ccb46c8a77134f1aa22ce76 : SUCCESS in 4m 16s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44c8553f3ab347fe8477ebfc7d2cf3ab : SUCCESS in 4m 27s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/263ffb4f07fd40f39ed7a7cd8ac42527 : SUCCESS in 31m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c004630e97841a8aa78deded3a19b2c : SUCCESS in 43m 36s\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e4031d675be44f6a8435f4632c2ef20 : SUCCESS in 32m 14s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/24d2e33dbb56429387d1abac1acb7e00 : SUCCESS in 40m 55s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ddfc2d4123b436fa26fb2ab95740751 : SUCCESS in 20m 17s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ea7bf7e4dac4681934e936b4ef872a4 : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/533c426cd6214efcbed777a38fd2560e : SUCCESS in 13m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ece863465e2a4b2bb11de884641e7b4d : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"23a090d5f527db13798a823ac61cef7b7c6308fe","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2025-06-23 06:20:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"544a96fc9f0dae5e2c60c217b2ef9feef294e4bb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-06-23 06:54:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0834dd1ecfe32b430ad3a00ea0fa707cbf1a01d5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-06-23 20:45:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1c1efab6da66b5df4dde95c9e072dc93b366167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 20:46:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"66fc97513d18bcd3cc33610b6de9e4ecdcc6efd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 21:20:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76ddde546c594e169c24704a9c8c0fe3\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/671e48b0b7ed418db32e1649f77ae905 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fd4438d44ae4175aeb45fae67df2fac : SUCCESS in 2m 05s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d164f373d4f45f2b7c3efa0d805a93c : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0a66107726c416fa1e3abfb6eebba1e : SUCCESS in 5m 00s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43681e02e4184510a643c4cc04a00a35 : SUCCESS in 5m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/76f57c84d7db42469470df403b0e46ff : SUCCESS in 30m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6f12e0b8af74cfaa145de1909ec660a : SUCCESS in 28m 11s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/25133685671a400fa4d2e47c74e1bc0f : SUCCESS in 25m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbd6056fb7204b56adf0253ef221b367 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e1b3a5ca8d23de04901c47f1939719aeace14bb3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 21:20:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"13ae089cb6e116f645bcf23b70171426d71632f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 21:21:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c69f5097ae91424bb35d1f797acdf72e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84510019c3984718836b858b4429aaa3 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c341ff213849962e4a807033bd02b53d003d2dc","revisions":{"7f875a025de9db5b9feb03e721bd91f9588f7d9e":{"kind":"REWORK","_number":1,"created":"2025-06-20 21:54:55.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/24/953024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/24/953024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/1"}}},"commit":{"parents":[{"commit":"c6fcfb83294b84acc419bd9edb0dde20232744c4","subject":"Merge \"Remove watch_log_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c6fcfb83294b84acc419bd9edb0dde20232744c4"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 21:54:28.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 21:54:28.000000000","tz":120},"subject":"Add queue_manager and stream_fanout","message":"Add queue_manager and stream_fanout\n\nThis patch adds two new parameters to the toplevel class:\n- rabbit_use_queue_manager\n- rabbit_stream_fanout\n\nthat are already handled by the oslo::messaging::rabbit class, which\nmakes it impossible to set using octavia_config.\n\nChange-Id: I75037abd5f0d17eb50d97b8355509c782325c788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7f875a025de9db5b9feb03e721bd91f9588f7d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7f875a025de9db5b9feb03e721bd91f9588f7d9e"}]},"branch":"refs/heads/master"},"7c341ff213849962e4a807033bd02b53d003d2dc":{"kind":"REWORK","_number":2,"created":"2025-06-20 23:07:36.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/24/953024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/24/953024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/24/953024/2"}}},"commit":{"parents":[{"commit":"c6fcfb83294b84acc419bd9edb0dde20232744c4","subject":"Merge \"Remove watch_log_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c6fcfb83294b84acc419bd9edb0dde20232744c4"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 21:54:28.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 23:07:26.000000000","tz":120},"subject":"Add queue_manager and stream_fanout","message":"Add queue_manager and stream_fanout\n\nThis patch adds two new parameters to the toplevel class:\n- rabbit_use_queue_manager\n- rabbit_stream_fanout\n\nthat are already handled by the oslo::messaging::rabbit class, which\nmakes it impossible to set using octavia_config.\n\nChange-Id: I75037abd5f0d17eb50d97b8355509c782325c788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7c341ff213849962e4a807033bd02b53d003d2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7c341ff213849962e4a807033bd02b53d003d2dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
