)]}'
{"id":"openstack%2Fpuppet-octavia~954617","triplet_id":"openstack%2Fpuppet-octavia~master~I07824ef319409c8611c9900b3c778d6ed1852f93","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-14 14:06:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07824ef319409c8611c9900b3c778d6ed1852f93","subject":"Add support for castellan library options","status":"MERGED","created":"2025-07-10 14:34:44.000000000","updated":"2025-07-14 14:07:11.000000000","submitted":"2025-07-14 14:06:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954617","meta_rev_id":"b66f77138d7c18cfe15b0a2390e8156d219245be","_number":954617,"virtual_id_number":954617,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-14 14:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-07-11 14:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-07-14 13:02:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-07-14 13:02:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 15:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-11 14:20:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-07-14 13:02:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8194229d41f401f954e291a9a12e5443d84212c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-10 14:34:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c54a028a984c134bad7edf914f8a8020db642ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:43:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793ce610978f456f94a6e0101af0d6d9\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02f0530e1c2740599fd838e45e61d2b3 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a42eb08bc2c4ca08bd824565b608e31 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/598df6f29a754ed8a941176621fbfd04 : SUCCESS in 3m 14s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/27af5d5e39ab4986bfed87f8ef27e655 : SUCCESS in 8m 52s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12a0fa1f42e249108ec3486e658c9266 : SUCCESS in 5m 08s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd949093ae914e2984d81ecec5426fd9 : SUCCESS in 5m 36s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed504b62c4d5407f920583ebe0af2364 : SUCCESS in 1h 01m 54s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c3e8557524244f3963f3a793a211ea3 : SUCCESS in 46m 37s\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bb75b05d5994ff28b15d3b68bc0bf12 : SUCCESS in 42m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29e4e1166133404fadfbcfd01295ee67 : SUCCESS in 53m 16s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd7732b67f4c4de7904c22a40bd4d0c6 : SUCCESS in 26m 54s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/135f3914ffbb49c58c5e12934b476132 : SUCCESS in 17m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f9c5696652457c95b58a6f7362995d : SUCCESS in 15m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be7f1a850914fd3acfd4d5af914eee7 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"07ca3fd4f67f91abc5f44ea7e86bf39f73abb675","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-07-11 14:20:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2fc512a9285d2655b831cb1b778073755d4493e1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-07-14 13:02:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dafaf8f14f7c5f795c91e82103b46907996c7ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 13:02:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f65649d6a93741ad23240a2aa5f5a63339fd20da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 14:06:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83e2e2a340294c4399e33dbaf95c49c3\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da98ea1b602348a4bc454e6f734bf2f7 : SUCCESS in 2m 13s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad5f9748afef453c8da804f055575cd2 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef504fca35e6429aa80f36185222cf95 : SUCCESS in 2m 25s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/bef52008b2964eceafcf6f7e4281c4f7 : SUCCESS in 9m 33s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec5db079c8dc4a1e983345e6c61a86b8 : SUCCESS in 5m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3358c458b4643468be33c1d2a5f0f9a : SUCCESS in 58m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/666d4058c4bc4859bbec62d76faef12f : SUCCESS in 52m 46s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/80affe7cd648481fb5384317b49d1c7b : SUCCESS in 21m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db11bff77c044f0b9623f499fa19e468 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f5d7ab8e7f53a2ca96daaf8375b5571aa99ea51f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 14:06:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b66f77138d7c18cfe15b0a2390e8156d219245be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 14:07:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2519a575118409b8f1a091f1131b904\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab51a4ba8a66484f91446bd810480a33 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"572baa50c62014420dacd73f0b7d207b50aa0fbe","revisions":{"572baa50c62014420dacd73f0b7d207b50aa0fbe":{"kind":"REWORK","_number":1,"created":"2025-07-10 14:34:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/954617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/17/954617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/954617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/954617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/954617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/17/954617/1"}}},"commit":{"parents":[{"commit":"7c341ff213849962e4a807033bd02b53d003d2dc","subject":"Add queue_manager and stream_fanout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7c341ff213849962e4a807033bd02b53d003d2dc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-10 14:26:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-10 14:34:36.000000000","tz":540},"subject":"Add support for castellan library options","message":"Add support for castellan library options\n\nThese options are used when castellan cert manager is used.\n\nChange-Id: I07824ef319409c8611c9900b3c778d6ed1852f93\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/572baa50c62014420dacd73f0b7d207b50aa0fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/572baa50c62014420dacd73f0b7d207b50aa0fbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
