)]}'
{"id":"openstack%2Fpuppet-octavia~958951","triplet_id":"openstack%2Fpuppet-octavia~master~Icd1b113f366a65a94a872262aca356ce0574f48b","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-01 02:59:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Icd1b113f366a65a94a872262aca356ce0574f48b","subject":"Fix wrong package name in parameter description","status":"MERGED","created":"2025-08-30 14:39:06.000000000","updated":"2025-09-01 03:54:10.000000000","submitted":"2025-09-01 03:54:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958951","meta_rev_id":"2fb10baa73112ea9bf0ed8b62d7f44257c05d6a7","_number":958951,"virtual_id_number":958951,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-01 03:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-09-01 02:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-08-31 12:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-09-01 02:59:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-30 15:37:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-31 12:13:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b56e740094eddad280bebad331719d1445e44684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-30 14:39:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17eab5a1d0a2af20e70976e0c4b1c1fe254f677a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 15:37:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00b07cc950e24488a939da5b1da7d3fa\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d907233da0724096a50c3a00d9030b47 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe216dbf21774ee78321ac0375f3a618 : SUCCESS in 2m 48s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/a70a9341b2ee4bcdb4a6dcafca09d864 : SUCCESS in 4m 13s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e23788e382d34cf9b6ac12e80d559504 : SUCCESS in 5m 59s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2441408edc614a5b931a96b0bfd15f31 : SUCCESS in 6m 05s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d9c1117d6e6473988d98114d0d716b7 : SUCCESS in 47m 04s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/194bbc9c63b8406bbf49e4069a6ed926 : SUCCESS in 53m 18s\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1457f5b8c7ae40c7bdee4c2f41585d3f : SUCCESS in 26m 34s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb17cc5b8b1c4b88bc433e95de857ab0 : FAILURE in 46m 10s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9b1aefbf27d4ca8bf01e4eec51f7a8a : SUCCESS in 28m 08s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/671828aeb4234d8ab101b32edc7805e6 : SUCCESS in 26m 35s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4989f3ff5fd34004bc079951a7946e48 : SUCCESS in 24m 05s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/180135e335414052a6ee75376940ce9b : SUCCESS in 9m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a480c9f3dca84e89b246646ba558d957 : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"851f304e761c86da5c5b0fefc57f8c178c4a34c4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-31 12:13:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ddce2bc2eb639c336114091e20efb932ccbeaea5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-01 02:59:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03c6846e359d9a4c9ef57328fadda5c010d04cda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 03:00:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"946d65e12c6684d9588c732e878838f75fbba683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 03:54:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95fa286c34834ffb95c27bdf1662af9e\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9996ec16142147959e655ad31540de8a : SUCCESS in 5m 07s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fbe9e916a63427c89d6bbf07d417d33 : SUCCESS in 5m 07s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/9e9d674dff0c4e5480b07babdc5e7064 : SUCCESS in 4m 16s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b245665fb7a84796934d5f78d4fae53a : SUCCESS in 5m 14s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18cf7f7c5b34467cb8aed49ab51d0d98 : SUCCESS in 48m 37s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22a9eb7a0c84472bb79e9202acb0083e : SUCCESS in 46m 53s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/32439686028640d9ab4409cf4ffe9a94 : SUCCESS in 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2fb10baa73112ea9bf0ed8b62d7f44257c05d6a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 03:54:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ba1524fb8fb417b226acd15f8929912630a459d","revisions":{"3ba1524fb8fb417b226acd15f8929912630a459d":{"kind":"REWORK","_number":1,"created":"2025-08-30 14:39:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/958951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/958951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/958951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/958951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/958951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/958951/1"}}},"commit":{"parents":[{"commit":"ebbbf82376dd836355a27a02ae8d996d33e58468","subject":"Add support for [DEFAULT] host option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ebbbf82376dd836355a27a02ae8d996d33e58468"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-30 14:38:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-30 14:38:30.000000000","tz":540},"subject":"Fix wrong package name in parameter description","message":"Fix wrong package name in parameter description\n\nChange-Id: Icd1b113f366a65a94a872262aca356ce0574f48b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3ba1524fb8fb417b226acd15f8929912630a459d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3ba1524fb8fb417b226acd15f8929912630a459d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
