)]}'
{"id":"openstack%2Fpuppet-openstack-integration~745894","triplet_id":"openstack%2Fpuppet-openstack-integration~stable%2Fqueens~I5acda250e2c134918d8c22e8b010e20cfc3f9b33","project":"openstack/puppet-openstack-integration","branch":"stable/queens","hashtags":[],"change_id":"I5acda250e2c134918d8c22e8b010e20cfc3f9b33","subject":"Force overwriting existing ssh keys","status":"MERGED","created":"2020-08-12 13:48:14.000000000","updated":"2020-08-13 01:23:51.000000000","submitted":"2020-08-13 01:23:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745894-1597281831992-ff0595dd","meta_rev_id":"79ef8379b824f49c506d59c3b63c1fd5e93d7636","_number":745894,"virtual_id_number":745894,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 01:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-12 15:25:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-13 00:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-13 00:13:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 15:25:56.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-13 00:13:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-13 01:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"909de1d9da5ef5de9241ce21f3d33c54ebdba1fc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-12 13:48:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2bcd4f7d157eb2ff125c73d04624d41b9c557a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 15:03:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/784cfc0babc54bdb83082504e4d2f7a2 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/a5426c042463464eb4c4703027543219 : SUCCESS in 5m 22s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/12de30794f42469492c3a005857a55d5 : SUCCESS in 1h 00m 56s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/84a9e5a467214c108a75f8b8cbb2e904 : SUCCESS in 54m 33s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bdd2ecbf229341cd8c507c7cc94b2227 : SUCCESS in 55m 48s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4e46bd1a6fc34d298d241935a5e2c718 : SUCCESS in 44m 25s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/1ad241a966694984b0dca30de89288f0 : FAILURE in 31m 59s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c0824a815a144daf8420f6af9de7398a : FAILURE in 27m 10s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/53438ad02f7c4d488bfea82e7e72eeb4 : FAILURE in 21m 47s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/a90a9d5bc61a4d7ba2ff0080914cbf4d : FAILURE in 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b00100bf182796f85fec5f7bcd05ba46ecea9599","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-12 15:25:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e58cfe8e47f84181ee036f7c3bdcb2926acc207","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-13 00:13:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41d0e1d73cd088850f5abe654475214fb570dacf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:14:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c63334eed3e03763287dda2d5bd491103c82a29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 01:23:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79ef8379b824f49c506d59c3b63c1fd5e93d7636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 01:23:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/6e8dac06cbfc450aa307523aaf3ac79b : SUCCESS in 3m 21s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/14685c214eba495f8afb79427dea6f7b : SUCCESS in 3m 30s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/d6a8c458df724f29bb155f3acdb7df76 : SUCCESS in 36m 50s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/37cfdf28924e4d458d79401954c1732f : SUCCESS in 38m 17s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a294b75f27124003b0ac5d5036ca0f1d : SUCCESS in 1h 08m 06s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1cbab3f906084d6b8ed7e9a908053133 : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32efae041d674d0f8fa88c1228ce4b67c7cf8336","revisions":{"32efae041d674d0f8fa88c1228ce4b67c7cf8336":{"kind":"REWORK","_number":1,"created":"2020-08-12 13:48:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/745894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/94/745894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/94/745894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/94/745894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/94/745894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/94/745894/1"}}},"commit":{"parents":[{"commit":"619bf96644f45071c2b146a12c692bafe20e471c","subject":"Ensure key format is PEM in beaker tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/619bf96644f45071c2b146a12c692bafe20e471c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-21 11:37:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-12 13:48:06.000000000","tz":540},"subject":"Force overwriting existing ssh keys","message":"Force overwriting existing ssh keys\n\nChange-Id: I5acda250e2c134918d8c22e8b010e20cfc3f9b33\n(cherry picked from commit 25bff52a31baf838ee60a7f2b3ad0aa0b622d008)\n(cherry picked from commit a37e1b99aedd2fae2ae80754802d7350bd835bc3)\n(cherry picked from commit a640aadc4bebac05e46f26da865b23a27431e7c0)\n(cherry picked from commit eaa2921e196688f0fde8404b02ae431509b731ce)\n(cherry picked from commit 5e745602674c35e6509cd5d9e1edefb03a933a37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/32efae041d674d0f8fa88c1228ce4b67c7cf8336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/32efae041d674d0f8fa88c1228ce4b67c7cf8336"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
