)]}'
{"id":"openstack%2Fpuppet-openstack-integration~876554","triplet_id":"openstack%2Fpuppet-openstack-integration~stable%2Ftrain~I2a5fb3692f09847540d3af94854387acea102430","project":"openstack/puppet-openstack-integration","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-07 08:13:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a5fb3692f09847540d3af94854387acea102430","subject":"Clone puppet-openstack_spec_helper during job","status":"MERGED","created":"2023-03-06 09:28:07.000000000","updated":"2023-03-07 08:13:12.000000000","submitted":"2023-03-07 08:13:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876554","meta_rev_id":"df51bfba3e3603413bb8460fbba1c199827e27f3","_number":876554,"virtual_id_number":876554,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 08:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-03-06 15:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-03-07 06:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-03-07 06:57:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 10:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 15:21:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-03-07 06:57:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b435c3f522d8d717ebe5a1f3e271a6fe80c49803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-06 09:28:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"760e2940603c595eed32389d36cedcfbe3aa4ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 10:36:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94c47c4f9c9b494d8f64673b94bf9e30\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65f84d96e64f4ff6a920b66a0621da2a : SUCCESS in 3m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08afec05a1944e65888c6fe564e0cc57 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eea4664dfe8542df908517faab28a243 : SUCCESS in 3m 47s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/6f331236cbe44cd690043b8b09591ed8 : SUCCESS in 53m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/42394a19f3184ba3b959ed95e36b064b : SUCCESS in 56m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7c486b57179345cfbedd6d530cf531f7 : SUCCESS in 58m 32s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/9acd5e745ede41cf884d383dc879b618 : SUCCESS in 48m 06s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9b84e846fbdc44c49c2bfa324fde78a7 : SUCCESS in 42m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b78f4547946406ab7b88f3dd456e3e4 : SUCCESS in 42m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4345c8cd6634a7ea1e9faeb6a2018b2 : SUCCESS in 32m 42s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5ede3398aadb4df490e34a52e264edcd : SUCCESS in 40m 04s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9f77e1842404933932830267b8ba2dc : SUCCESS in 52m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdfc1b15ec9e41c99a4e4845d83fd4f9 : SUCCESS in 1h 00m 55s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad7ca32224aa4df982b7936db520253c : SUCCESS in 54m 36s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28666e2e006041a9aa2fcbfeb6c03bf4 : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c4a2a7c88981cd59c9b578a27e98f90aa4492f15","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-06 15:21:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28094e9e68cab84025d677b4c3b34fd469aa6356","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-07 06:57:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bed658c9f4f30c7b8fcec0ddc8043749ebe0d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 06:58:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a302c44157aa7ead6bfb8feea70b9c6f83cfea0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 08:13:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de1aca9da4b4607a5c6b4c98e6d791b\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9403853a70fa45cd9fa49f1662960f97 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc0ac59a0fd546f085053ab1c6a070a2 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07127a99e4c94d88b6eecb454241ba42 : SUCCESS in 4m 16s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/19232f76339f483ab4e91e7ed33d074c : SUCCESS in 55m 58s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e72c50b7781f4cd3be5c3c10c14f2774 : SUCCESS in 1h 03m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c0a1729d5ff14542a8852eb6cc29c3b0 : SUCCESS in 1h 10m 02s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/899a1264327444e5bf74542bf3864248 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"df51bfba3e3603413bb8460fbba1c199827e27f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 08:13:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8441433aba48c3d069e86981d1d0237feb2b3bd","revisions":{"e8441433aba48c3d069e86981d1d0237feb2b3bd":{"kind":"REWORK","_number":1,"created":"2023-03-06 09:28:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/876554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/54/876554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/54/876554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/54/876554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/54/876554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/54/876554/1"}}},"commit":{"parents":[{"commit":"37cf0054a5478f0a920710838eda8e23553a8619","subject":"TripleO: Remove unused project template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/37cf0054a5478f0a920710838eda8e23553a8619"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-12 03:59:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-06 09:27:56.000000000","tz":540},"subject":"Clone puppet-openstack_spec_helper during job","message":"Clone puppet-openstack_spec_helper during job\n\nCurrently we are supposed to use cloned puppet-openstack_spec_helper\nto run some jobs, thus we should add it to required projects.\n\nConflicts:\n\tzuul.d/base.yaml\n\nChange-Id: I2a5fb3692f09847540d3af94854387acea102430\n(cherry picked from commit cff4d9f2b2626f9cb29145c5c038bdb0e045479d)\n(cherry picked from commit cfc63dae37b9df9169b26a4b6016c76e321a38b2)\n(cherry picked from commit 63d5e50803438e90a0c410929b926f131c8e4811)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/e8441433aba48c3d069e86981d1d0237feb2b3bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/e8441433aba48c3d069e86981d1d0237feb2b3bd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
