)]}'
{"id":"openstack%2Fpuppet-openstack-integration~918167","triplet_id":"openstack%2Fpuppet-openstack-integration~unmaintained%2Fwallaby~Iaa284b7e6a157ca532f263b2229bec2dff3da2f2","project":"openstack/puppet-openstack-integration","branch":"unmaintained/wallaby","topic":"WX_unmaintained_change","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-12-16 06:41:09.000000000","reason":"Change was abandoned"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-12-16 06:41:09.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-27 01:21:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"34750":{"account":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"last_update":"2024-12-16 06:41:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaa284b7e6a157ca532f263b2229bec2dff3da2f2","subject":"Pull puppet openstack modules from unmaintained/wallaby","status":"ABANDONED","created":"2024-05-06 08:02:06.000000000","updated":"2024-12-16 06:41:09.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":916308,"cherry_pick_of_patch_set":2,"meta_rev_id":"da87a44b1b8e40d9c78032591469a0aacda6411c","_number":918167,"virtual_id_number":918167,"owner":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-26 21:53:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-06-24 00:55:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-06-24 00:55:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-06 09:03:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-21 13:05:25.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2024-05-21 13:06:04.000000000","updated_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-05-28 06:06:02.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-05-30 14:30:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"},{"updated":"2024-06-24 00:55:45.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-12-16 06:41:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"83d98c762edfc86c1c5db1acd7663b02b5e5b197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-06 08:02:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"53016ebae7c718f43305e8e8c65818fd34eab2b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 09:03:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20ec2b42954640dc9e7a4c5b1079392e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/244d61be4e244da78a1e9df873359ee8 : SUCCESS in 15m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f90568e6e834374a208ff13d289f2eb : SUCCESS in 7m 47s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6486ba05a2f435cb78db2e770bc837e : FAILURE in 6m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23e0d0e37e064ff0a947743c6e7e6b5d : FAILURE in 7m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4686a57d08fd4e299aa21fdf14ca14d4 : FAILURE in 6m 11s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a254acdad524758ada58672e028c0da : FAILURE in 5m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7982a30d9924982b2e998b098e897a7 : FAILURE in 13m 54s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2005a9cd14924068a449afaa7b4ba2fc : FAILURE in 13m 44s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/00ed3079da6e43b8a22bf9c780f98fa8 : FAILURE in 13m 23s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba4a8b83cd2c4856994fe6c63b608ad0 : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":1},{"id":"95cdbe0f47b7dc9f88a22e020a5ec0557508b8b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-07 14:17:17.000000000","message":"Topic WX_unmaintained_change-unmaintained/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"bb33ba206e2cb2bf5c9e1ee600d9a941e5c59df6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-07 14:17:21.000000000","message":"Topic set to WX_unmaintained_change","accounts_in_message":[],"_revision_number":1},{"id":"54098609625148f2588a562fdd1c47d3bad585b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-20 13:01:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c853bc443e238319c21e735e8deba54d0be40cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 13:01:39.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/puppet-openstack-integration on branch unmaintained/wallaby.  The\nproblem was:\n\n  The project \"openstack/puppet-panko\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"puppet-openstack-integration-base\" job stanza:\n\n  job:\n      name: puppet-openstack-integration-base\n      abstract: true\n      parent: puppet-openstack-base\n      required-projects:\n        - openstack/openstack\n        - name: openstack/puppet-aodh\n          override-checkout: unmaintained/wallaby\n        - name: openstack/puppet-barbican\n  ...\n\n  in \"openstack/puppet-openstack-integration/zuul.d/base.yaml@unmaintained/wallaby\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"a0159428c5e1acc4c6d7e5afa0d0c7bc7a088480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-20 13:16:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b279e44c6b3d830d4a313d099ab9b76b76787426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 13:16:53.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/puppet-openstack-integration on branch unmaintained/wallaby.  The\nproblem was:\n\n  The project \"openstack/puppet-panko\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"puppet-openstack-integration-base\" job stanza:\n\n  job:\n      name: puppet-openstack-integration-base\n      abstract: true\n      parent: puppet-openstack-base\n      required-projects:\n        - openstack/openstack\n        - name: openstack/puppet-aodh\n          override-checkout: unmaintained/wallaby\n        - name: openstack/puppet-barbican\n  ...\n\n  in \"openstack/puppet-openstack-integration/zuul.d/base.yaml@unmaintained/wallaby\", line 2","accounts_in_message":[],"_revision_number":3},{"id":"72c98754ef64523e6b1fcd4f2a556f9c29bb33cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-20 13:29:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d6e4453aa31687685743c080fff634d920fae5f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 14:48:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff8c90945ff149a494f18bf124f52542\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fd35fe7d8cd45008fca6c7bc9b7459c : SUCCESS in 18m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a24ccaf2e89b43bc894c1de686b7722b : SUCCESS in 9m 27s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74d9318642df4b5e9325e4dd9e4281fc : FAILURE in 5m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeec510535c344718025ba9000ff26a2 : FAILURE in 6m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e27dbbddebcb4f1f9903d4046d01b863 : FAILURE in 6m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f74f4287ab994c878cfc55277a7ba517 : FAILURE in 6m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5fa087958514823a6b0e3319bd59cb1 : FAILURE in 13m 19s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/98607bbc28394bc99abd9eccfc79228a : SUCCESS in 1h 11m 36s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8fc6c12047d46319e539faa13afc40b : FAILURE in 14m 07s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2a7b71666134e8eaf245ca7777e71e1 : SUCCESS in 51m 58s","accounts_in_message":[],"_revision_number":4},{"id":"2b5901bda27ce1c70efd6c3b0c11b5a6f1923b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-20 15:02:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b566e09577703a5867f9d7753ebbd2dbdfe521aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:24:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bb71533737241708ef82b5db9358a46\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c7c2be2689a4ecd9131357d8d766a7e : SUCCESS in 17m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27f59e7df4264013ad2f4985cc40fdfa : SUCCESS in 9m 10s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81d4ce7bde0147d9ad0ae4dfa3f04f1f : FAILURE in 8m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3af3639a35804fef8ca126e125df65d8 : FAILURE in 6m 20s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f98792ff750f433f95219b73b6ec7bae : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c130c469c0046f3b9bbdb3cce68f575 : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/89545f6585b1430daf55321f8ab8e8b9 : FAILURE in 13m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8cf0276225e04e12b968772b2fd9d3e7 : SUCCESS in 1h 21m 33s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc0fa634c2484de0a81c5c5d37fa181e : FAILURE in 21m 32s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a139eea3ecb347eca8658d7779eabbc8 : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":5},{"id":"1992c83da8e201c525414580bb83365fa20d7d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-21 05:35:18.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c1a0a02c2922b54fe4c7b6ec17110cae285904e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 06:55:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/748a70dc57764a4f8e2e7e4c42f83ec9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d64ce0f54fc349fa81c730564d534fa0 : SUCCESS in 17m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f0d553b130c42c18d08b13de82a340a : SUCCESS in 8m 23s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf89ac7ebed453a96b9950f87c163d3 : FAILURE in 6m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38fca04b03964e67aba2afc4411d24f7 : FAILURE in 6m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/883d31162f9c4277b80bc62be9fb1d07 : FAILURE in 6m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/722ad94ee5f1494c96decbe494e681ed : FAILURE in 6m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b473ffb4e3846e2b8254fc05e1c47f4 : FAILURE in 16m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/758443d27e104d59993762dd8ef4c46c : SUCCESS in 1h 08m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a88723c216848bdb6622e1975d035a3 : SUCCESS in 1h 15m 00s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4362a3b29484929b9fb69a3d1ad86b2 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":6},{"id":"81e6485f2dbf5c00680cd5abffa6480ac2ced0a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-21 07:28:47.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d6cdf6fd35f029c9f9a4e64e6e46c9fda31be513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 09:33:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c12dbb8710c485b827580d15e0d4458\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4a55aabc0664f27be0a21210583e8ee : SUCCESS in 15m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4cfa2c87fe84cddb4e8383233a57d84 : SUCCESS in 8m 26s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c771cb6e3202480f8878f51e1aa102f2 : FAILURE in 6m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08bb61990bda4a3dae09e0b5c164eb60 : FAILURE in 6m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4b67db8f7ee4e8f8a272b90da79b31d : FAILURE in 6m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e15efabc38d44fe0b297108b7d783bee : FAILURE in 6m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8bfb365d6794e4a9e8b427723dd7097 : TIMED_OUT in 2h 02m 40s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1ac2f67d34140989b2062df0df43dd2 : SUCCESS in 1h 08m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad1f1134072f490698372ee9c84ebe3b : SUCCESS in 1h 14m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1321680b9b874e1c92e926522c90b39f : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":7},{"id":"735985b88224ff2dc62eaa2235275e6ae080c6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-21 11:02:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fdeaf76869d590d93abd36fbd3e70dce068daf17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 12:27:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b4451a70f274c40abf2a2cab24cc406\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fba7c46ebf944358cd769a7b5d4089c : SUCCESS in 17m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd63540cb041436b93e76fb94fe5804c : SUCCESS in 8m 53s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f3692d87efe43ddbaa1d338bff0cdc0 : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26e20ad2bc1a48ee920f93d8674143cb : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/482e6eb233244e44b24a72ceaa628a64 : FAILURE in 8m 10s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbe83ad935cd4dd594b5bb553d81ae39 : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fd6ec081a57432bb0092b76cfed7dad : SUCCESS in 1h 17m 40s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99c9cb3f9dc444149b2db0e8ac96112f : SUCCESS in 1h 15m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c53759627c654f7cb5d6426e1acf4138 : SUCCESS in 1h 09m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2500aa4045ac4a26b10d2e4592a25fba : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":8},{"id":"1d54452bfe96f09fc4a27610da16faef4a830992","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-05-21 13:05:25.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8f510b10995477e262a76f73ea1e81850728e83a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-21 13:08:45.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"c7ded1a182679baa5f25af95417e0197d3dd6b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 14:43:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e9f579f71644618f21945b6a695281\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26b6155c7baf4ac3a20b7d448c1393bd : SUCCESS in 19m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f198f13312294e709a09679343f1d0e3 : SUCCESS in 8m 07s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eca8dd448b94e299825be122012ce9a : FAILURE in 6m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88f7cf71eb1c4b38992295439d58f710 : FAILURE in 5m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f976995bd6f4215a0e98f9a873788a3 : FAILURE in 8m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1c05c3f7ea84d8d9d43e5bb44a69217 : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/81e60849fbee493a9fada1c16a9f8abd : SUCCESS in 1h 13m 48s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8aaab30cdb2445328d3604096cde3e82 : SUCCESS in 1h 28m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f03f248e619f43c59116388836f019ef : SUCCESS in 1h 11m 28s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1f7956bd5b94126994ff9a4c025ef0c : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":9},{"id":"c6a4d5ae652e051e4b710b81910d65b2a6edfac1","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-05-21 14:53:00.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"086b283bd6b747e418bd1e9dc4daf2a4343646ba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-05-28 11:05:52.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fdfc6d18c36f0052218f696565136a8aae51342b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-29 10:04:17.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5fc193d171b5566eab402043f58a6c140c4414d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-29 10:20:31.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"8dba483a5e6d3538e134c12c62a14372a4d5dce2","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-29 10:20:36.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7aedc456d6145e5fbdf6b69eed64c9804ecfe28f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 10:49:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcd630d70dde48fe91aa5de9d45905da\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b3d118657a94815983b13d5116c01a9 : SUCCESS in 16m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9444cb36465415ea5d04e71319a5762 : SUCCESS in 6m 26s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9f3e1985f2541239e281e4e683bb01c : FAILURE in 8m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac6bb3b5fd684fcfaba88c26f5b269c9 : FAILURE in 5m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30b5cd1fdd1c4ab4bd52b93ee28898c7 : FAILURE in 5m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/811f8bf12b164754a5d470a824e8d530 : FAILURE in 6m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b94fdb9ad06a44cbbc3ab759d349a60a : FAILURE in 15m 26s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5a630f181424f348f2842c96b9e8b7d : FAILURE in 12m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/57aa0c54fe25438c950e1350212f2128 : FAILURE in 20m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3c212244a274f729a21e3f8d4666075 : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":10},{"id":"12a6188e27a9430c4a54142b470302e0b4df4a97","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-05-29 14:55:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0354dd17f57f78bb31631ff5711744a355e99e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 15:28:14.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7547bb9a6f5d4b55a7ca1272f2043496\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc9541c08d884b40a7661e9e441e2429 : SUCCESS in 18m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ab8232e031e4de4b5f7bac2d537e2a0 : SUCCESS in 10m 20s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6df12c9b1ae43e1bfb072c3c54abee3 : FAILURE in 6m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fb5202a3ac449b8954e1b2035456f72 : FAILURE in 6m 16s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bccaefcb2134b049844bff9cc3e87b7 : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/721ef01ae0144045a756987ab39d39bd : FAILURE in 7m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35d970a16d1a403da1e37560d298520c : FAILURE in 15m 54s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cdbf048968f45fba5adc7c8bd2a4acb : FAILURE in 12m 39s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7849ceeac45e4991878313846476e380 : FAILURE in 14m 19s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cfbcdc5e64f43c8a8fe2659fd191912 : FAILURE in 15m 31s","accounts_in_message":[],"_revision_number":10},{"id":"cc409d33a7a8380b4285696782573fe184ec55a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-05-31 07:06:12.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"c5ee0feea37d3bfb354283cb966428088892ba64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 08:34:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e5515c0d587472685aaef6332037496\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94f2d3979fa5468abbd8995cc89b148c : SUCCESS in 17m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/640597489b924cee857f09293111ee9a : SUCCESS in 7m 37s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bccfdf1c83ca4433aae382327a9e0e4b : FAILURE in 6m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a652d8857c30464191f50cb14b597d8a : FAILURE in 10m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce941cd75a264ced8a3bbbc7ce14a05a : FAILURE in 7m 40s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e5c28c9a7b5493686247d3444183ca2 : FAILURE in 7m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/636cf1cdb57442d6a7a2989ffd433b41 : SUCCESS in 1h 03m 39s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/696b788a990c4fd080caacfa430bd30d : SUCCESS in 1h 00m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d45662f40adc4084884a6a60ec4d37bf : SUCCESS in 1h 00m 16s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9cef56d811449eebe5227ae7c53399f : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":11},{"id":"bca6476980bf6938fa56ce948f950e94f5e9ebba","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-24 00:55:45.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"9b7139e70eab46236074146dbdd6e47abeaadada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 00:56:59.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4c8301be32132f90e3223a11a452250061c2b3a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 02:02:43.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ab89a37e4ec48db85dc7778f8c0108a\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a8c8a15448443ab9d18b286909049e8 : SUCCESS in 16m 13s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65bbf0f88df54e8bbfc5dc61a63e7621 : SUCCESS in 7m 52s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/555373715986431994730e88238b8bd5 : RETRY_LIMIT in 4m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8013c839cd2f46e99b1bd047959fabf0 : RETRY_LIMIT in 3m 57s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e83937853a6e4f38a49c269e9fda1b65 : RETRY_LIMIT in 4m 01s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bb562d0f83f4702becdc55c47e2c437 : RETRY_LIMIT in 4m 09s","accounts_in_message":[],"_revision_number":11},{"id":"22bb407c905152f9466f8465ec97e3958757116a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-11-26 21:53:14.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"e0d81d60456865f114bcf941a5312713c77f8c3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 21:53:37.000000000","message":"Patch Set 11: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/puppet-openstack-integration on branch unmaintained/wallaby.  The\nproblem was:\n\n  The nodeset \"centos-8-stream\" was not found.\n\nThe problem appears in the the \"puppet-openstack-integration-6-scenario001-tempest-centos-8-stream\" job stanza:\n\n  job:\n      name: puppet-openstack-integration-6-scenario001-tempest-centos-8-stream\n      parent: puppet-openstack-integration-6-scenario001\n      nodeset: centos-8-stream\n\n  in \"openstack/puppet-openstack-integration/zuul.d/integration.yaml@unmaintained/wallaby\", line 23","accounts_in_message":[],"_revision_number":11},{"id":"2cc4e4e5253df13a5694b86a5d468ffab479c2b5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-27 01:21:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dd01f0f273e364ef9f2496e5da9006bc504e68b5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-11-27 14:49:36.000000000","message":"Patch Set 11:\n\nah i didn\u0027t notice this was xena/wallaby. yea we have a handful of reviews still outstanding for these that likely need to be abandoned at this point","accounts_in_message":[],"_revision_number":11},{"id":"da87a44b1b8e40d9c78032591469a0aacda6411c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-16 06:41:09.000000000","message":"Abandoned\n\nunmaintained/wallaby branch of openstack/puppet-openstack-integration transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"99d1d325a29cbb72c92e0f89f482a59004b48310","revisions":{"e838516b311c1900225ada59338654d111537485":{"kind":"REWORK","_number":1,"created":"2024-05-06 08:02:06.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/1"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"Karolina Kula","email":"kkula@redhat.com","date":"2024-05-06 08:02:06.000000000","tz":0},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/e838516b311c1900225ada59338654d111537485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/e838516b311c1900225ada59338654d111537485"}]},"branch":"refs/heads/unmaintained/wallaby"},"80aaa957ec974404a215b9cfe715ad2c6b23cdc0":{"kind":"REWORK","_number":2,"created":"2024-05-20 13:01:07.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/2"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-20 13:00:59.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/80aaa957ec974404a215b9cfe715ad2c6b23cdc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/80aaa957ec974404a215b9cfe715ad2c6b23cdc0"}]},"branch":"refs/heads/unmaintained/wallaby"},"ae9159f50991ac619d9ce734dbb79f0c649efbf4":{"kind":"REWORK","_number":3,"created":"2024-05-20 13:16:23.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/3"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-20 13:16:13.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/ae9159f50991ac619d9ce734dbb79f0c649efbf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/ae9159f50991ac619d9ce734dbb79f0c649efbf4"}]},"branch":"refs/heads/unmaintained/wallaby"},"5022efdd61af5f81d41f2ff6daba3662d0417652":{"kind":"REWORK","_number":4,"created":"2024-05-20 13:29:24.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/4"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-20 13:29:15.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/5022efdd61af5f81d41f2ff6daba3662d0417652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/5022efdd61af5f81d41f2ff6daba3662d0417652"}]},"branch":"refs/heads/unmaintained/wallaby"},"47d30aaccce1142df92b21137e5f69a5206f5cdc":{"kind":"REWORK","_number":5,"created":"2024-05-20 15:02:11.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/5"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-20 14:05:05.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/47d30aaccce1142df92b21137e5f69a5206f5cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/47d30aaccce1142df92b21137e5f69a5206f5cdc"}]},"branch":"refs/heads/unmaintained/wallaby"},"8d5e1c5b611da577cef86df72c015e58640093b9":{"kind":"REWORK","_number":6,"created":"2024-05-21 05:35:18.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/6"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-21 05:34:34.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/8d5e1c5b611da577cef86df72c015e58640093b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/8d5e1c5b611da577cef86df72c015e58640093b9"}]},"branch":"refs/heads/unmaintained/wallaby"},"aad4c2f4f93fb3338dccc9d4e81042c2a0748a01":{"kind":"REWORK","_number":7,"created":"2024-05-21 07:28:47.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/7"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-21 07:28:34.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/aad4c2f4f93fb3338dccc9d4e81042c2a0748a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/aad4c2f4f93fb3338dccc9d4e81042c2a0748a01"}]},"branch":"refs/heads/unmaintained/wallaby"},"376cd782b65d13c5f014ddea8ced04d99e41b73a":{"kind":"REWORK","_number":8,"created":"2024-05-21 11:02:41.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/8"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-21 11:02:33.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/376cd782b65d13c5f014ddea8ced04d99e41b73a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/376cd782b65d13c5f014ddea8ced04d99e41b73a"}]},"branch":"refs/heads/unmaintained/wallaby"},"6b0aa4f873f9883b84204eca4c428fe448d5f403":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-05-21 13:08:45.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/9"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"Karolina Kula","email":"kkula@redhat.com","date":"2024-05-21 13:08:45.000000000","tz":0},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nThis change also required removing usage of retired modules:\n- panko\n- ec2api\n- sahara\n- amqp\n- qdr\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/6b0aa4f873f9883b84204eca4c428fe448d5f403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/6b0aa4f873f9883b84204eca4c428fe448d5f403"}]},"branch":"refs/heads/unmaintained/wallaby","description":"Edit commit message"},"41adbbbd383c7416e05334ea51962967cc82586d":{"kind":"REWORK","_number":10,"created":"2024-05-29 10:20:31.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/67/918167/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/10"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-05-29 10:20:20.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nThis change also required removing usage of retired modules:\n- panko\n- ec2api\n- sahara\n- amqp\n- qdr\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/41adbbbd383c7416e05334ea51962967cc82586d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/41adbbbd383c7416e05334ea51962967cc82586d"}]},"branch":"refs/heads/unmaintained/wallaby"},"99d1d325a29cbb72c92e0f89f482a59004b48310":{"kind":"REWORK","_number":11,"created":"2024-05-31 07:06:12.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/67/918167/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/67/918167/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/67/918167/11"}}},"commit":{"parents":[{"commit":"3cef1a78c1efde2db3e37fe23aa8253cf2e704e4","subject":"Undefine ssl certificate parameters for mariadb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/3cef1a78c1efde2db3e37fe23aa8253cf2e704e4"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-04-18 15:20:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-05-31 07:06:10.000000000","tz":120},"subject":"Pull puppet openstack modules from unmaintained/wallaby","message":"Pull puppet openstack modules from unmaintained/wallaby\n\nBranch name has moved from stable/wallaby to unmaintained/wallaby in all\npuppet modules as part of switch to unmaintained phase for wallaby.\n\nThis change also required removing usage of retired modules:\n- panko\n- ec2api\n- sahara\n- amqp\n- qdr\n\nChange-Id: Iaa284b7e6a157ca532f263b2229bec2dff3da2f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/99d1d325a29cbb72c92e0f89f482a59004b48310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/99d1d325a29cbb72c92e0f89f482a59004b48310"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
