)]}'
{"id":"openstack%2Fpuppet-openstack-integration~996029","triplet_id":"openstack%2Fpuppet-openstack-integration~master~I8e022d1e1ebe1b49c612e92e343ed832a13c2383","project":"openstack/puppet-openstack-integration","branch":"master","topic":"remove-postgresql","hashtags":[],"change_id":"I8e022d1e1ebe1b49c612e92e343ed832a13c2383","subject":"Remove puppetlabs-postgresql","status":"MERGED","created":"2026-07-05 03:52:05.000000000","updated":"2026-07-05 06:59:46.000000000","submitted":"2026-07-05 06:59:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996029-remove-postgresql","meta_rev_id":"a3b31c471db7c64c35697d41eebed9534dd8dfd1","_number":996029,"virtual_id_number":996029,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-05 06:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-07-05 05:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-07-05 05:34:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-05 05:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fb33a11ffada8dd0abaca18671a860c5a7abc87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-05 03:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5305121ea70ffd77e37df33145f1e9269960642c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 05:26:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d9db49611b248709b7303593b89dee8\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ea988ac2351482d9ad34f360aa741a4 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c868a4af0f3947b5a71ef4ee5e0634e6 : SUCCESS in 2m 47s\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f289b69aca90435caa13238e1e801ac6 : SUCCESS in 28m 59s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f82eb6ae59744ef186178e904cbeb93b : FAILURE in 40m 50s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9ebcc092c5f498c8513e9f1a41ef2c1 : FAILURE in 1h 02m 31s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ce50e19a1be4ea6bffbbdbf1c36d206 : SUCCESS in 54m 07s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00f99e1e59f5477e86a6f31c9e5ee862 : FAILURE in 52m 10s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aabdd4dd77824ace9f0d51b16125bfc5 : SUCCESS in 34m 07s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/884472c09b1f41a9a6a7e1a0eec77ac3 : SUCCESS in 41m 16s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/920d0bcfabe94ae0a2553b9ada8fd26b : SUCCESS in 27m 07s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20eb2b89e59e49bab27439a51700a974 : SUCCESS in 57m 03s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5406f3ca0df542fdbbf8df1272d24773 : SUCCESS in 26m 44s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcaca8065c854a23b68f0b699731f318 : SUCCESS in 30m 08s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c0e2a364378466fb4f09404b2b4cb40 : SUCCESS in 1h 28m 05s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4c8de6cf9b644308b705bb722e5352d : SUCCESS in 30m 16s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1ebae11281f4a09b6f776c39c4fbec6 : SUCCESS in 42m 11s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2471a89a868b4969920b602bbb36e24c : SUCCESS in 1h 00m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fe17f4a64d89b17e126ee64fa28d62df4befa7e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-05 05:34:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"581335aca5c03eb11707073f52a343bd373ce7a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 05:34:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e241167b45e24d8ef1f09b7ac73ae55262685927","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 06:59:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac2e541a1d234d51843aba4a244cfa6e\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5907535ffc9a43bcbfbdfff6f924441b : SUCCESS in 3m 07s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b3d03a0de594171ad9b07475d7783d7 : SUCCESS in 1m 52s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/361d70dc8c6840b79672e35e10fe0dbc : SUCCESS in 48m 35s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51f23bfb22084fc29ed20b870c69f7a2 : SUCCESS in 1h 19m 41s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01c98d4f4b044dc79b57d9a80bf1d7c1 : SUCCESS in 31m 13s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c54024f55b64f4bb5793632d452d7cc : SUCCESS in 27m 25s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f7b903ad688492ca694e090f77dd831 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a3b31c471db7c64c35697d41eebed9534dd8dfd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 06:59:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f22ea473fb0956e6817c673edb993cdfefda7667","revisions":{"f22ea473fb0956e6817c673edb993cdfefda7667":{"kind":"REWORK","_number":1,"created":"2026-07-05 03:52:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/996029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/29/996029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/29/996029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/29/996029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/29/996029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/29/996029/1"}}},"commit":{"parents":[{"commit":"12f7128f9bf1cf11cca8b1be68f2e9231fda346f","subject":"Updated from Puppet OpenStack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/12f7128f9bf1cf11cca8b1be68f2e9231fda346f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-05 03:45:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-05 03:51:48.000000000","tz":540},"subject":"Remove puppetlabs-postgresql","message":"Remove puppetlabs-postgresql\n\nThe module is no longer used.\n\nDepends-on: https://review.opendev.org/993266\nDepends-on: https://review.opendev.org/993267\nChange-Id: I8e022d1e1ebe1b49c612e92e343ed832a13c2383\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/f22ea473fb0956e6817c673edb993cdfefda7667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/f22ea473fb0956e6817c673edb993cdfefda7667"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
