)]}'
{"id":"openstack%2Fpuppet-openstack-integration~996047","triplet_id":"openstack%2Fpuppet-openstack-integration~master~I4cabc475ae33bbb886217e85c89fe7135d236f45","project":"openstack/puppet-openstack-integration","branch":"master","hashtags":[],"change_id":"I4cabc475ae33bbb886217e85c89fe7135d236f45","subject":"Revert \"Use \u0027proposed\u0027 packages\"","status":"NEW","created":"2026-07-05 15:51:13.000000000","updated":"2026-07-06 01:02:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa889183a106bac32d81612f00d9ba11d384aa49","_number":996047,"virtual_id_number":996047,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-05 17:22:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-07-06 01:02:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-05 17:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6d7dcb1130161237cb1285f2b2d96e978f783b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-05 15:51:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2c6cf7c34c8e47ea190187b539333eae91b5d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 17:22:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de3a7cdfe0004c83b239d8166070ca3b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e552e5f93b34d03b13ffbf18d63dd94 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/795a542b34174e5c8472bfaf484c5430 : SUCCESS in 1m 49s\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6db919a313934e628f7033c4f3d51e3a : SUCCESS in 1h 04m 18s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c82be5f62124f2aa4d8fa341df76535 : FAILURE in 1h 13m 38s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/807ea2fd510148459f5c5a28d139c03f : FAILURE in 1h 19m 21s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f26a04ed6ec4997bef12298dd66b78a : SUCCESS in 50m 47s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b979e82e202742b7b442a1665f40a918 : FAILURE in 54m 20s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7183a8cfa5f8453aa44e3b752129e702 : SUCCESS in 1h 05m 56s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72a4f95a07564fb5964136450c90041f : SUCCESS in 46m 09s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7af893969ea143259374f84c5e4d2a00 : SUCCESS in 32m 06s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc36e94c58514b569632e5da2f99c28e : SUCCESS in 25m 46s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c6ac024bade4f1b8bf9cf0a325ffbfa : SUCCESS in 56m 41s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ba03af163464b1e85d858f42bd0b098 : SUCCESS in 1h 06m 03s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1125a59981834573b6074ce3202abbaa : SUCCESS in 1h 25m 05s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4616d0bbf532465db9e0a131fd3e5ae1 : SUCCESS in 1h 04m 15s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4dc9e109a98451b833092aff10058dd : SUCCESS in 54m 58s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d4e688840d74b33ad1f35787af0f64f : SUCCESS in 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa889183a106bac32d81612f00d9ba11d384aa49","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-06 01:02:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2229b2948e07ac87c2b2c42491e375c241eb34c9","revisions":{"2229b2948e07ac87c2b2c42491e375c241eb34c9":{"kind":"REWORK","_number":1,"created":"2026-07-05 15:51:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/996047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack-integration","ref":"refs/changes/47/996047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/47/996047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/47/996047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/47/996047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack-integration refs/changes/47/996047/1"}}},"commit":{"parents":[{"commit":"12f7128f9bf1cf11cca8b1be68f2e9231fda346f","subject":"Updated from Puppet OpenStack modules constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/12f7128f9bf1cf11cca8b1be68f2e9231fda346f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-05 15:49:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-05 15:50:59.000000000","tz":540},"subject":"Revert \"Use \u0027proposed\u0027 packages\"","message":"Revert \"Use \u0027proposed\u0027 packages\"\n\nThis reverts commit fcf9b3aba7ba4a2826a23f82724bfd024e7e96ce.\n\nReason for revert:\nNow the required versions are all available in \u0027updates\u0027 packages.\n\nConflicts:\n\tconfigure_facts.sh\n\nChange-Id: I4cabc475ae33bbb886217e85c89fe7135d236f45\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/2229b2948e07ac87c2b2c42491e375c241eb34c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack-integration/commit/2229b2948e07ac87c2b2c42491e375c241eb34c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
