)]}'
{"id":"openstack%2Fpuppet-openstack_spec_helper~1004433","triplet_id":"openstack%2Fpuppet-openstack_spec_helper~master~I408ffe578191106a4626f8988da1f17fc27fc7af","project":"openstack/puppet-openstack_spec_helper","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-07 14:52:54.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I408ffe578191106a4626f8988da1f17fc27fc7af","subject":"DNM: Complete migration to openvox","status":"NEW","created":"2026-09-07 14:37:02.000000000","updated":"2026-09-07 14:52:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cec5ec957fb4a70a98746616471d0fc712a24dea","_number":1004433,"virtual_id_number":1004433,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-07 14:52:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 14:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edfe59f1bca46916472f297d1e9dd88edb0266a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 14:37:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cec5ec957fb4a70a98746616471d0fc712a24dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:52:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10f6be9f8dee44a0938f8574dcca95e4\n\n- puppet-openstack-libraries-puppet-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f7cd78c903d4dafaed521d053a42d65 : FAILURE in 2m 54s\n- puppet-openstack-libraries-puppet-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c920f917999540bf8f4ca4a412292814 : FAILURE in 4m 00s\n- puppet-openstack-libraries-puppet-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/4642c5e12fef4c20be6f5069c9d8666a : FAILURE in 5m 55s\n- puppet-openstack-libraries-puppet-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9a50c8ecb3a4b1daa4264ecb6a4ddf8 : FAILURE in 6m 18s\n- puppet-openstack-libraries-puppet-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e48e75d4875402085a969da2cf725c2 : FAILURE in 3m 28s (non-voting)\n- puppet-openstack-libraries-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/85e3c6b84dc742f3a7a6ab5224340836 : FAILURE in 8m 57s\n- puppet-openstack-libraries-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcd557ade85141cebcea9b6fd700be36 : FAILURE in 5m 46s (non-voting)\n- puppet-openstack-libraries-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b42b1d2d215b42f7b6f376bb61a10989 : FAILURE in 6m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/289c26013f2643e494a474a71e3c9d0b : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98b88c1b372dff181b2d2340ed03708870c73f4e","revisions":{"98b88c1b372dff181b2d2340ed03708870c73f4e":{"kind":"REWORK","_number":1,"created":"2026-09-07 14:37:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/1004433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack_spec_helper","ref":"refs/changes/33/1004433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/33/1004433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/33/1004433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/33/1004433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/33/1004433/1"}}},"commit":{"parents":[{"commit":"779375305802632453199bdfe0f90ad6bb068752","subject":"Add voxpupuli-test and voxpupuli-acceptance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/779375305802632453199bdfe0f90ad6bb068752"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 14:36:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 14:36:43.000000000","tz":540},"subject":"DNM: Complete migration to openvox","message":"DNM: Complete migration to openvox\n\nChange-Id: I408ffe578191106a4626f8988da1f17fc27fc7af\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/98b88c1b372dff181b2d2340ed03708870c73f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/98b88c1b372dff181b2d2340ed03708870c73f4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
