)]}'
{"id":"openstack%2Fpuppet-openstack_spec_helper~707993","triplet_id":"openstack%2Fpuppet-openstack_spec_helper~master~I7eb23eef7f9ed199c2c28863ef04d64f6acfd644","project":"openstack/puppet-openstack_spec_helper","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I7eb23eef7f9ed199c2c28863ef04d64f6acfd644","subject":"[ussuri][goal] Cleanup for python 2.7 drop","status":"MERGED","created":"2020-02-16 02:23:18.000000000","updated":"2020-02-18 03:21:47.000000000","submitted":"2020-02-18 03:20:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707993-1581996022799-18a4250f","meta_rev_id":"1e1114c96a58eec7269252e7ecd52245a0d01ef5","_number":707993,"virtual_id_number":707993,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 03:20:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-02-18 02:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-02-17 20:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-02-18 02:42:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 20:32:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-18 02:42:41.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-02-18 03:20:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c716c2625957e182f6e073748ccebaa036ad0905","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-16 02:23:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"484bc46291917d2920bc6c9afc87cfaeb2ae1e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 02:57:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-libraries-puppet-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d64d206beba94f90aabe891f5e3445d6 : SUCCESS in 7m 14s\n- puppet-openstack-libraries-puppet-syntax-5-centos-7 https://zuul.opendev.org/t/openstack/build/54e8c4d49bb84c35bcf5f62e2264b19f : SUCCESS in 11m 35s\n- puppet-openstack-libraries-puppet-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c0855af3b7444f783fe645408152ab8 : SUCCESS in 7m 41s\n- puppet-openstack-libraries-puppet-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4a96cf4d4d8e47faa782c1ecbcbbc239 : SUCCESS in 23m 58s\n- puppet-openstack-libraries-puppet-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0baea9a5b5284e928a4de93203986285 : SUCCESS in 18m 57s\n- puppet-openstack-libraries-puppet-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/868f2e8f995a4e8aa2634bb3807089d0 : FAILURE in 16m 22s (non-voting)\n- puppet-openstack-libraries-puppet-beaker-rspec-centos-7 https://zuul.opendev.org/t/openstack/build/5ebf5df52bec4c9389f8db15da8d5884 : SUCCESS in 29m 00s\n- puppet-openstack-libraries-puppet-beaker-rspec-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/715812b0aa4b4f6699840677604cde05 : FAILURE in 22m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d23b49bad4d94455b5ea70055211695e : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1370af9d511439de956f1ffc26fc9dabdcd15f50","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:32:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6842c41ade4fec9fdfc53025fb0829404f5e391","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-18 02:42:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cadeee745fd34afdc765ad265d97f0dc436a73d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 02:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d21cd113b0b8d1a9571eeb5ccdacf73fc551e3dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:20:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2f8c0789f284a3069aa97e7c19f425a132377448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:20:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-libraries-puppet-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bcc1328717b4d72bcc8a521d7d0318c : SUCCESS in 8m 23s\n- puppet-openstack-libraries-puppet-syntax-5-centos-7 https://zuul.opendev.org/t/openstack/build/4053c57587d342fcbe17f19eea5c8717 : SUCCESS in 11m 53s\n- puppet-openstack-libraries-puppet-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26435ac403ef433aa5afb7e57f0d3182 : SUCCESS in 10m 09s\n- puppet-openstack-libraries-puppet-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/94365a5b118b4b89b88553866f263f95 : SUCCESS in 29m 37s\n- puppet-openstack-libraries-puppet-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d508d388e69140e5ac7e9da404f90ae3 : SUCCESS in 18m 57s\n- puppet-openstack-libraries-puppet-beaker-rspec-centos-7 https://zuul.opendev.org/t/openstack/build/421a86ed45ef47279e92bcfe221756e9 : SUCCESS in 34m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10d64d54b9444e3e8bc76f9f902bd571 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9a1f20cfbc12cd320f94ea236d1e3950ceb9449b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:21:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f6eebb8c29d45d78b534696f4cb07b8 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"881bd9d2f94624ca7d1b4d922798538ed165bd15","revisions":{"881bd9d2f94624ca7d1b4d922798538ed165bd15":{"kind":"REWORK","_number":1,"created":"2020-02-16 02:23:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/707993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack_spec_helper","ref":"refs/changes/93/707993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/93/707993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/93/707993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/93/707993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/93/707993/1"}}},"commit":{"parents":[{"commit":"57785448ce4e0d2ffece98022c268c9e773d3912","subject":"Pin gettext \u003c 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/57785448ce4e0d2ffece98022c268c9e773d3912"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 02:22:05.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 02:23:17.000000000","tz":-360},"subject":"[ussuri][goal] Cleanup for python 2.7 drop","message":"[ussuri][goal] Cleanup for python 2.7 drop\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\npuppet-* repos have to cleanup the requirement and tox\nfor py2.7 drop.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I7eb23eef7f9ed199c2c28863ef04d64f6acfd644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/881bd9d2f94624ca7d1b4d922798538ed165bd15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/881bd9d2f94624ca7d1b4d922798538ed165bd15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
