)]}'
{"id":"openstack%2Fpuppet-openstack_spec_helper~776580","triplet_id":"openstack%2Fpuppet-openstack_spec_helper~stable%2Ftrain~I277eafe7a3ba5392090945f07ef2ebca71f6035e","project":"openstack/puppet-openstack_spec_helper","branch":"stable/train","topic":"train-test","hashtags":[],"change_id":"I277eafe7a3ba5392090945f07ef2ebca71f6035e","subject":"DNR/DNM quick test","status":"ABANDONED","created":"2021-02-19 07:12:23.000000000","updated":"2021-02-19 16:16:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"688e8cb0469c6c6e781fdea320ea4549d6d0925e","_number":776580,"virtual_id_number":776580,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-19 07:44:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 07:44:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc7c1c0d2906c72523bc8e7b7e226134067ef6f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-02-19 07:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d6e8817bbb8c1222ce776235d258f0024cc9bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 07:44:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-libraries-puppet-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfac546d8d6d44739bb0c88417081283 : SUCCESS in 7m 04s\n- puppet-openstack-libraries-puppet-syntax-5-centos-7 https://zuul.opendev.org/t/openstack/build/a5350d55028c45e79ce84beaaf8cb315 : FAILURE in 9m 03s\n- puppet-openstack-libraries-puppet-syntax-5-centos-8 https://zuul.opendev.org/t/openstack/build/31511fed7e9b4bffa93e30782e25d3c0 : SUCCESS in 11m 50s\n- puppet-openstack-libraries-puppet-syntax-6-centos-8 https://zuul.opendev.org/t/openstack/build/d77576de72124de1ba80632edf4df242 : SUCCESS in 11m 06s\n- puppet-openstack-libraries-puppet-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95eca248e3394665956e385770167833 : SUCCESS in 7m 28s\n- puppet-openstack-libraries-puppet-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2cf7e7941a6f4d8098769e8439d870b8 : FAILURE in 9m 09s\n- puppet-openstack-libraries-puppet-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0150c992ebaf490aa2f74ae040347bf8 : SUCCESS in 13m 32s (non-voting)\n- puppet-openstack-libraries-puppet-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/b09cdf7396484d60a558790a2b5a8de3 : SUCCESS in 13m 41s (non-voting)\n- puppet-openstack-libraries-puppet-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2acee1467454356bb88f8a1d1d9a2f6 : SUCCESS in 10m 39s\n- puppet-openstack-libraries-puppet-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82e7ffd48fa0474fa059ba7359c8c016 : SUCCESS in 25m 57s (non-voting)\n- puppet-openstack-libraries-puppet-beaker-rspec-centos-7 https://zuul.opendev.org/t/openstack/build/898cc2311ef24c5bb9b7424de546a759 : SUCCESS in 20m 17s\n- puppet-openstack-libraries-puppet-beaker-rspec-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d60310355f245bcbb8c8d30f5503866 : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"688e8cb0469c6c6e781fdea320ea4549d6d0925e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-02-19 16:16:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a6dc59f574ae244579c7c252f240df85b467275","revisions":{"8a6dc59f574ae244579c7c252f240df85b467275":{"kind":"REWORK","_number":1,"created":"2021-02-19 07:12:23.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/80/776580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-openstack_spec_helper","ref":"refs/changes/80/776580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/80/776580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/80/776580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/80/776580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-openstack_spec_helper refs/changes/80/776580/1"}}},"commit":{"parents":[{"commit":"30465d33207dcfb568a2be67b8c849ae7a6699b3","subject":"Pin beaker-hiera to 0.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/30465d33207dcfb568a2be67b8c849ae7a6699b3"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2021-02-19 07:11:31.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2021-02-19 07:11:38.000000000","tz":60},"subject":"DNR/DNM quick test","message":"DNR/DNM quick test\n\nTesting if the puppet-openstack-libraries-puppet-syntax-5-centos-7 job\nis working or not\n\nChange-Id: I277eafe7a3ba5392090945f07ef2ebca71f6035e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/8a6dc59f574ae244579c7c252f240df85b467275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-openstack_spec_helper/commit/8a6dc59f574ae244579c7c252f240df85b467275"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
