)]}'
{"id":"openstack%2Fpuppet-oslo~1004735","triplet_id":"openstack%2Fpuppet-oslo~master~I97edb9cf40b28c4424442126c11f4a582298f420","project":"openstack/puppet-oslo","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 10:03:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I97edb9cf40b28c4424442126c11f4a582298f420","subject":"Deprecate support for [database] mysql_enable_ndb","status":"MERGED","created":"2026-09-09 02:32:33.000000000","updated":"2026-09-10 10:04:51.000000000","submitted":"2026-09-10 10:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004735","meta_rev_id":"f86665bfac3db44b55f47e1ae666629d4ca22fc6","_number":1004735,"virtual_id_number":1004735,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 10:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2026-09-09 14:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2026-09-10 08:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2026-09-10 08:35:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 04:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 14:28:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2026-09-10 08:35:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4638893689c95aa1e784f692076d31245360f2ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13ad5067e7bbb116cef242c8107a9c834474660b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 04:46:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cf41062e4e74653a40722367333e211\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2fb3b52c7f446daa5fa5d01ffd0151c : SUCCESS in 3m 09s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25bacbec5c1a488daaa6be2b940ae49b : SUCCESS in 3m 20s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f664fba21f64420a85696685ebfa64a8 : FAILURE in 7m 09s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2868dc757f8d4afd9a6334b087cc07ba : FAILURE in 7m 42s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/976c1504f28640f5976e76e065825a81 : FAILURE in 6m 13s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/744d519bf9964440807bef1d116a9847 : SUCCESS in 1h 03m 14s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90bd35cf7c004f119b84eebe9dcb84ce : FAILURE in 1h 11m 40s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcd7c9ade26c424c8f07cba85fdf8e83 : FAILURE in 1h 28m 25s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0d37c0cf1674fde96245450e486a035 : SUCCESS in 45m 13s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0c48005d7fe44c4a498afd51f2cc8f4 : FAILURE in 55m 44s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ad75bf34b614519907503a88da76e4c : SUCCESS in 1h 12m 59s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35936e2efb564936aaa2836ef45a1add : SUCCESS in 1h 35m 33s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a24896bbeeb6434ca1db5e31505b68b7 : SUCCESS in 1h 03m 10s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/540354513936434b9681a0d500bcf96e : SUCCESS in 1h 08m 00s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57156853b42c43939e2bedcdd92da703 : SUCCESS in 1h 01m 20s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c636e5b1d1c4277a44c0317c9649665 : SUCCESS in 54m 57s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/278f92a80e034060987ffcfbbc20733a : SUCCESS in 42m 22s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e38e07b9e6284013800bc392a859c155 : SUCCESS in 1h 08m 55s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/917abec3a1da4dc2a3ee1d2afa08348e : SUCCESS in 1h 07m 53s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74ea21cdcf64462280a040b09417067d : SUCCESS in 1h 01m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b63996537f54f4a883a160c61a13fdb : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"44a2c5354319588416a36c3cf35ffd77cc0febfc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 09:44:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"686b1ebd09bc58a839a49f179e9bf176cce50347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 11:19:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9929aef04624afca6ab4fb9b7058eaa\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4ebe78f6b5a443781a4cd8e4ac6ebb0 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6f175a169d546958773113486b0ef02 : SUCCESS in 2m 32s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/95aa4bb829b74882b8e241f5d4b450e5 : SUCCESS in 6m 59s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60acedd805c84469b3aba9b5577195ca : SUCCESS in 5m 35s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d9642306714444a848781d8a776b2a3 : SUCCESS in 5m 32s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37c9c48e1eb542b6a5f8fba56ef6fdcd : SUCCESS in 1h 09m 19s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97784f1a3ebc455d88360a8c90aafd0b : FAILURE in 1h 09m 26s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53f8ffd4f9544f0caedd7a0ab59ef0d4 : FAILURE in 1h 21m 03s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ffd37b2e0ba40b2bc442749b97cf4be : SUCCESS in 51m 40s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fda0c645b1c445a68143c95d2623891a : FAILURE in 39m 13s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc5fdadad68246d585903768728eecb9 : SUCCESS in 1h 06m 58s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/391b171aacde49348e7b08340e9faf7e : SUCCESS in 49m 43s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a7806814a0c4ea59934f6a5aaec5188 : SUCCESS in 58m 29s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/223cb08420774145826dcc1123232744 : SUCCESS in 47m 04s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6af837110a2c47dc9bd8f02b6c51c9e1 : SUCCESS in 34m 41s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3da1de3d7cf54e518f9ec36ea5d68020 : SUCCESS in 59m 58s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/138f8d88c3684cc3898189a7e6437692 : SUCCESS in 1h 26m 15s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03774471ac8b4a14bd6e59179b9b7416 : SUCCESS in 30m 20s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c3c75e3089d48dbb8e2029acc8870ff : SUCCESS in 31m 49s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3beb526e36ea4541ad7ec68e3a63df83 : SUCCESS in 54m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f57c659910249f69aa77f352375e8f6 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cd7618a01664b8bc251e97b3b1d7533627fb4b96","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-09-09 14:28:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"168cbcda965f68398ad351d1e646c3560055871c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-10 08:35:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29261c4981530f0329a874a53d2b98e724145a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 08:37:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"361f9f05dea03bba8b50942807a4c0f7487bc6b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 10:03:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/706ecc1af37b4be0b6f73b569e214ad4\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cce2787440b04ca491d73d638e0d2e0a : SUCCESS in 3m 25s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51a9d1fe0c7e40cc867638cd3b65282e : SUCCESS in 2m 44s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/7870aad3d89043b0853abae9f03c600b : SUCCESS in 6m 21s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3242cfeb88b435b9c09f16d9c710b30 : SUCCESS in 6m 55s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ff0a5a3910c4b65bf0ea3db931a98df : SUCCESS in 1h 09m 47s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ac7bdfccb6a45f282b8d1af2fbe97da : SUCCESS in 1h 21m 35s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f780819613ae453f8972717ddd19c7aa : SUCCESS in 1h 03m 16s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f44a484954d4a839e0f4bec1b5c3a50 : SUCCESS in 57m 41s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f9cd3a70adf4f3b8c1c393a78381ae9 : SUCCESS in 59m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efcd3564e3744ad09c83737a0261919c : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1c68dec951f457b7d7739cfd69ae8cfb0ab054f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 10:03:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f86665bfac3db44b55f47e1ae666629d4ca22fc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 10:04:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e594e6337b4e769588fa6b1c45077f\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bbe3719543a4a5bb640eeb90b88321a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e127719158adaf21121a713642fd6ae603f0f60","revisions":{"1e127719158adaf21121a713642fd6ae603f0f60":{"kind":"REWORK","_number":1,"created":"2026-09-09 02:32:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/1004735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/35/1004735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/35/1004735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/35/1004735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/35/1004735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/35/1004735/1"}}},"commit":{"parents":[{"commit":"bb284ccbde0c433c2cf964cada24311934ffebda","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/bb284ccbde0c433c2cf964cada24311934ffebda"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:31:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:31:45.000000000","tz":540},"subject":"Deprecate support for [database] mysql_enable_ndb","message":"Deprecate support for [database] mysql_enable_ndb\n\nThis option was already removed from oslo.db a few years ago.\n\nDepends-on: https://review.opendev.org/889079\nChange-Id: I97edb9cf40b28c4424442126c11f4a582298f420\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1e127719158adaf21121a713642fd6ae603f0f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1e127719158adaf21121a713642fd6ae603f0f60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
