)]}'
{"id":"openstack%2Fpuppet-oslo~602852","triplet_id":"openstack%2Fpuppet-oslo~master~Iced735d76b3bdaef795faf9599da453a20b091f1","project":"openstack/puppet-oslo","branch":"master","topic":"bug/1792719","hashtags":[],"change_id":"Iced735d76b3bdaef795faf9599da453a20b091f1","subject":"Provide more useful fail message","status":"MERGED","created":"2018-09-15 15:19:55.000000000","updated":"2019-04-25 16:09:16.000000000","submitted":"2019-04-25 16:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"602852-1556208556846-0411a076","meta_rev_id":"416a7c0ef8ed2ae3cdbebc3823d8422354a2a7f5","_number":602852,"virtual_id_number":602852,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":2,"date":"2019-04-25 16:09:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-01-02 10:08:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2018-10-08 15:51:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2018-09-15 16:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2018-10-08 10:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-04-25 14:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2018-09-27 09:18:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2018-10-01 06:49:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-04-25 14:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-15 16:06:19.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2018-09-27 09:18:50.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2018-10-01 06:49:18.000000000","updated_by":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"reviewer":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"state":"REVIEWER"},{"updated":"2018-10-08 10:48:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2018-10-08 15:51:04.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-01-02 10:08:37.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-25 14:53:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-25 16:09:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c14793c5203be3d77579eaa4b87c981ddf7533c0","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2018-09-15 15:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac5d42c7c051f925bb41f681817e53d5d7ae29ac","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2018-09-15 15:52:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9653e21de5764a670c32f71cf38e7cea80642163","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2018-09-15 16:06:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3bbed25455d7061805491eed9a94231493eaa677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 17:08:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/52/602852/2/check/puppet-openstack-lint/cbee3b9/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-4 http://logs.openstack.org/52/602852/2/check/puppet-openstack-syntax-4/92154a1/ : SUCCESS in 3m 54s\n- puppet-openstack-syntax-5 http://logs.openstack.org/52/602852/2/check/puppet-openstack-syntax-5/693d7eb/ : SUCCESS in 3m 41s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-unit-4.8-centos-7/1b32d64/ : SUCCESS in 9m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-unit-5.5-centos-7/50e7f29/ : SUCCESS in 8m 43s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-unit-latest-centos-7/68daaa4/ : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/c6d7c0e/ : SUCCESS in 49m 58s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/58913b6/ : SUCCESS in 1h 05m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/4b06b92/ : SUCCESS in 1h 00m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/4eebcf5/ : FAILURE in 1h 10m 02s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/2070a1e/ : SUCCESS in 48m 56s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/23b775e/ : SUCCESS in 48m 43s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/6d18e8d/ : SUCCESS in 47m 00s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/dbca658/ : SUCCESS in 49m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/a75732f/ : SUCCESS in 43m 12s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/13e7d5b/ : SUCCESS in 54m 15s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/3158b87/ : SUCCESS in 51m 13s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/311e650/ : SUCCESS in 46m 52s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/95ea292/ : SUCCESS in 57m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-beaker-centos-7/9bf7b8f/ : SUCCESS in 7m 33s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/52/602852/2/check/puppet-openstack-beaker-ubuntu-bionic/017b02d/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"30d2dcc9ca80b8709f915320aff8bd570741b270","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2018-09-15 22:15:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a5ca0cac0898d9cad8928280d9e6c84c71f4fe51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 23:24:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/52/602852/2/check/puppet-openstack-lint/d7c8460/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-4 http://logs.openstack.org/52/602852/2/check/puppet-openstack-syntax-4/280f2fe/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-5 http://logs.openstack.org/52/602852/2/check/puppet-openstack-syntax-5/680760e/ : SUCCESS in 4m 31s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-unit-4.8-centos-7/e0b4521/ : SUCCESS in 9m 45s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-unit-5.5-centos-7/8d3813a/ : SUCCESS in 9m 26s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-unit-latest-centos-7/c024561/ : SUCCESS in 9m 58s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/49626a3/ : SUCCESS in 49m 45s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/a2f3420/ : SUCCESS in 58m 40s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/babd96d/ : SUCCESS in 57m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/7335ecf/ : SUCCESS in 1h 07m 14s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/919311b/ : SUCCESS in 51m 05s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/66e092b/ : SUCCESS in 47m 25s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/5e26430/ : SUCCESS in 45m 32s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/4c29ca4/ : FAILURE in 51m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/5047fd8/ : SUCCESS in 37m 26s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/3e22a75/ : FAILURE in 1h 02m 08s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/370abf4/ : SUCCESS in 54m 25s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/11aa24c/ : SUCCESS in 50m 22s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/52/602852/2/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/d11dfcf/ : SUCCESS in 47m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/52/602852/2/check/puppet-openstack-beaker-centos-7/aaa935b/ : SUCCESS in 8m 43s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/52/602852/2/check/puppet-openstack-beaker-ubuntu-bionic/8cfc41c/ : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"91375a8ff762e43feca19eec2053df7b9300a9b6","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-09-27 09:18:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4f867539890d47fab7381fa50fb982350cb734e1","author":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"date":"2018-10-01 06:49:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3b16fc30659090f639c448a2b40fcfca409d3e02","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-08 10:48:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f8590dff1dd5008d809e71e6b838bdc1d105dd03","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2018-10-08 15:51:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"959b4d7dc5436511395506ee65b0bd08d1bb2fa3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-01-02 10:08:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8440a14b64c1446217321a2fd31acd7051892b05","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-25 14:53:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b46e3823ba21c0ab7ad5c25960c8a4ba5613e73a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:53:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a8f70e46a936e358d8fc82fff69d0951134ba03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:09:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"95596c029e3890fa8b536ca3c8af0c0d23ce0279","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:09:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/52/602852/2/gate/puppet-openstack-lint-ubuntu-bionic/b0822d2/ : SUCCESS in 5m 50s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/52/602852/2/gate/puppet-openstack-syntax-4-ubuntu-bionic/13bf5b0/ : SUCCESS in 5m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/52/602852/2/gate/puppet-openstack-syntax-5-ubuntu-bionic/37ce25f/ : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/52/602852/2/gate/puppet-openstack-syntax-6-ubuntu-bionic/1191539/ : SUCCESS in 3m 48s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/52/602852/2/gate/puppet-openstack-unit-4.8-centos-7/3163608/ : SUCCESS in 16m 38s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/52/602852/2/gate/puppet-openstack-unit-5.5-centos-7/774878b/ : SUCCESS in 15m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/52/602852/2/gate/puppet-openstack-unit-6.0-ubuntu-bionic/fd093c4/ : SUCCESS in 11m 07s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/52/602852/2/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/d3ff8cf/ : SUCCESS in 55m 44s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/52/602852/2/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/19dd05f/ : SUCCESS in 1h 01m 11s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/52/602852/2/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/99d6cbe/ : SUCCESS in 1h 04m 43s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/52/602852/2/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/b582dfa/ : SUCCESS in 45m 18s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/52/602852/2/gate/puppet-openstack-beaker-centos-7/b13e5c1/ : SUCCESS in 9m 01s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/52/602852/2/gate/puppet-openstack-beaker-ubuntu-bionic/0cd572e/ : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83ea4f1f4534a18e115153c48c903c51876f1fa7","revisions":{"1d9e87b9667976bce0173fc0f86c65047400a7bf":{"kind":"REWORK","_number":1,"created":"2018-09-15 15:19:55.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/52/602852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/52/602852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/1"}}},"commit":{"parents":[{"commit":"041ac686c275f4a726ef3cf303debb62706ae8f0","subject":"Merge \"Add the missing \u0027Table of Contents\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/041ac686c275f4a726ef3cf303debb62706ae8f0"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:10:25.000000000","tz":-240},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:10:30.000000000","tz":-240},"subject":"Provide more useful fail message","message":"Provide more useful fail message\n\nChange-Id: Iced735d76b3bdaef795faf9599da453a20b091f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1d9e87b9667976bce0173fc0f86c65047400a7bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1d9e87b9667976bce0173fc0f86c65047400a7bf"}]},"branch":"refs/heads/master"},"83ea4f1f4534a18e115153c48c903c51876f1fa7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-09-15 15:52:03.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/52/602852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/52/602852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/52/602852/2"}}},"commit":{"parents":[{"commit":"041ac686c275f4a726ef3cf303debb62706ae8f0","subject":"Merge \"Add the missing \u0027Table of Contents\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/041ac686c275f4a726ef3cf303debb62706ae8f0"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:10:25.000000000","tz":-240},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:42:23.000000000","tz":-240},"subject":"Provide more useful fail message","message":"Provide more useful fail message\n\nChange-Id: Iced735d76b3bdaef795faf9599da453a20b091f1\nCloses-Bug: #1792719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/83ea4f1f4534a18e115153c48c903c51876f1fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/83ea4f1f4534a18e115153c48c903c51876f1fa7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
