)]}'
{"id":"openstack%2Fpuppet-oslo~654210","triplet_id":"openstack%2Fpuppet-oslo~master~If00cebdf0f0df9ce664ea19e2939f9eaa30bbc33","project":"openstack/puppet-oslo","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"If00cebdf0f0df9ce664ea19e2939f9eaa30bbc33","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:46:14.000000000","updated":"2019-04-22 18:18:31.000000000","submitted":"2019-04-22 18:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654210-1555957111493-354c625b","meta_rev_id":"d8a911c30b375a4764da85e63fd29df013824ddc","_number":654210,"virtual_id_number":654210,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 18:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-04-22 13:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 17:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 17:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:23:12.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 17:05:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 18:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c2b721a97ce492267b461a4b012bc6263560713","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:46:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7fc37b8bd8457f8e193e2e4aac1aac70c84d1c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 08:11:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-lint-ubuntu-bionic/861eafc/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-syntax-4-ubuntu-bionic/221d372/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-syntax-5-ubuntu-bionic/4093ebe/ : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-syntax-6-ubuntu-bionic/e9a69c3/ : SUCCESS in 3m 41s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/654210/1/check/puppet-openstack-unit-4.8-centos-7/3896f45/ : SUCCESS in 16m 15s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/654210/1/check/puppet-openstack-unit-5.5-centos-7/b35e47d/ : SUCCESS in 14m 33s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/9edee13/ : SUCCESS in 8m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-unit-latest-ubuntu-bionic/8e5ab78/ : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/98bb9c6/ : SUCCESS in 1h 00m 16s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/10ec7a0/ : SUCCESS in 1h 00m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/5eb84a7/ : SUCCESS in 58m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/c356f2f/ : SUCCESS in 42m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/2f9ea5c/ : FAILURE in 56m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/bb66629/ : FAILURE in 30m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/063abc9/ : FAILURE in 15m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/10/654210/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/723e75e/ : FAILURE in 33m 23s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/10/654210/1/check/puppet-openstack-beaker-centos-7/2f39c37/ : SUCCESS in 8m 07s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/10/654210/1/check/puppet-openstack-beaker-ubuntu-bionic/4bf1229/ : SUCCESS in 9m 01s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/10/654210/1/check/build-openstack-releasenotes/8cb840b/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0beb53fdbf31f37b03335cae5b7add189b2ff7a8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:23:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc25038c35e1662b10df4e0c6c93dd7c0c6121d1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 17:05:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4bb4ead58765594bd32ffb430349d8fac568e9a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:05:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"620808e4ddc9fefd2528cb66c16536804d859347","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:18:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da3dd932f1a480b0dca66b62b7bb579a47327ef5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:18:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/654210/1/gate/puppet-openstack-lint-ubuntu-bionic/d5ff490/ : SUCCESS in 4m 55s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/654210/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/3c869e5/ : SUCCESS in 4m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/654210/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/58ec95b/ : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/654210/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/ea15c28/ : SUCCESS in 5m 00s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/654210/1/gate/puppet-openstack-unit-4.8-centos-7/835e53e/ : SUCCESS in 9m 45s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/654210/1/gate/puppet-openstack-unit-5.5-centos-7/ebce66d/ : SUCCESS in 10m 05s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/10/654210/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/33709bf/ : SUCCESS in 7m 38s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/10/654210/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/88cf4e9/ : SUCCESS in 1h 03m 04s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/10/654210/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/392c767/ : SUCCESS in 1h 00m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/10/654210/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/19746e2/ : SUCCESS in 56m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/10/654210/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/500d24a/ : SUCCESS in 45m 03s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/10/654210/1/gate/puppet-openstack-beaker-centos-7/60d2419/ : SUCCESS in 8m 35s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/10/654210/1/gate/puppet-openstack-beaker-ubuntu-bionic/5fbce8a/ : SUCCESS in 6m 05s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/10/654210/1/gate/build-openstack-releasenotes/e583e44/html/ : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd18dd9454b8488763d410341aa8d142eee7b17f","revisions":{"dd18dd9454b8488763d410341aa8d142eee7b17f":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:46:14.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/10/654210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/654210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/654210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/654210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/654210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/654210/1"}}},"commit":{"parents":[{"commit":"8be52339a33e8eef577aeb511d1b874130997ddc","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/8be52339a33e8eef577aeb511d1b874130997ddc"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: If00cebdf0f0df9ce664ea19e2939f9eaa30bbc33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/dd18dd9454b8488763d410341aa8d142eee7b17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/dd18dd9454b8488763d410341aa8d142eee7b17f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
