)]}'
{"id":"openstack%2Fpuppet-oslo~663192","triplet_id":"openstack%2Fpuppet-oslo~master~Iafde83ea8a92758d46164d4b177d502df883a6bc","project":"openstack/puppet-oslo","branch":"master","topic":"prep-release-train-1","hashtags":[],"change_id":"Iafde83ea8a92758d46164d4b177d502df883a6bc","subject":"Prepare Train M2","status":"MERGED","created":"2019-06-05 01:42:55.000000000","updated":"2019-06-10 13:28:13.000000000","submitted":"2019-06-10 13:28:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663192-1560173293012-811cd442","meta_rev_id":"53e1b98e87ad70efa9fbd292e777f65cb55af261","_number":663192,"virtual_id_number":663192,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 13:28:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-06-10 13:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-06-10 13:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 13:21:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 13:28:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8858517fa702f7da2a71cedf2262fb79caf576c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8ad8c970d419fa094bbccfea28533ee2ad00eae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:17:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/92/663192/1/check/build-openstack-puppet-tarball/5ba4a7a/ : SUCCESS in 1m 59s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/92/663192/1/check/puppet-openstack-lint-ubuntu-bionic/053509d/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3535c384fc8d2b5185e97a99d7138aad3c72138a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-10 02:09:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e396e1619717d2f971c64df8356cca4424fc8a99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:18:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/92/663192/2/check/build-openstack-puppet-tarball/eb64875/ : SUCCESS in 1m 56s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/92/663192/2/check/puppet-openstack-lint-ubuntu-bionic/a11d34f/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d9808275198aba81659969bf0500283b0463e885","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:21:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a2fea6cb8056a333b60baeab584f086a8c38e173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:22:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce7b526c8b8c8422df2d6c2e27444e32c5ba11ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:28:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/92/663192/2/gate/build-openstack-puppet-tarball/c3cb4eb/ : SUCCESS in 2m 09s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/92/663192/2/gate/puppet-openstack-lint-ubuntu-bionic/4316ae8/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"bf3fe622c2353eeccb2adfd55d63a74e517410d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:28:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81289178e6f29dfa3bebed4d0b8feeb1062c654c","revisions":{"bd8b822f4e6701cb812f2baf9af6980da530ece0":{"kind":"REWORK","_number":1,"created":"2019-06-05 01:42:55.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/92/663192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/92/663192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/1"}}},"commit":{"parents":[{"commit":"908503e3e972aa17220d2814762d77f91c689052","subject":"Prefix memcached_servers IPv6 address with `inet6:`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/908503e3e972aa17220d2814762d77f91c689052"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:21.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:21.000000000","tz":480},"subject":"Prepare Train M1","message":"Prepare Train M1\n\nUpdate the version metadata for Train milestone 1\n\nChange-Id: Iafde83ea8a92758d46164d4b177d502df883a6bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/bd8b822f4e6701cb812f2baf9af6980da530ece0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/bd8b822f4e6701cb812f2baf9af6980da530ece0"}]},"branch":"refs/heads/master"},"81289178e6f29dfa3bebed4d0b8feeb1062c654c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 02:09:23.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/92/663192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/92/663192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/92/663192/2"}}},"commit":{"parents":[{"commit":"908503e3e972aa17220d2814762d77f91c689052","subject":"Prefix memcached_servers IPv6 address with `inet6:`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/908503e3e972aa17220d2814762d77f91c689052"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:21.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-10 02:05:01.000000000","tz":480},"subject":"Prepare Train M2","message":"Prepare Train M2\n\nUpdate the version metadata for Train milestone 2\n\nChange-Id: Iafde83ea8a92758d46164d4b177d502df883a6bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/81289178e6f29dfa3bebed4d0b8feeb1062c654c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/81289178e6f29dfa3bebed4d0b8feeb1062c654c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
