)]}'
{"id":"openstack%2Fpuppet-oslo~672441","triplet_id":"openstack%2Fpuppet-oslo~master~Id1ae971d8f099b770841db382269f5ed039d8069","project":"openstack/puppet-oslo","branch":"master","topic":"prep-release-train-3","hashtags":[],"change_id":"Id1ae971d8f099b770841db382269f5ed039d8069","subject":"Prepare Train M3","status":"MERGED","created":"2019-07-24 08:20:27.000000000","updated":"2019-07-24 13:45:03.000000000","submitted":"2019-07-24 13:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672441-1563975903423-53055c7d","meta_rev_id":"7f1971b02a0eb9347d0b4ab0e6b4ecb98b1a1417","_number":672441,"virtual_id_number":672441,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-24 13:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-07-24 13:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-07-24 13:32:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 13:32:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-24 13:45:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63aabf8e73aaa8b48f07998700117311c6e54b11","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-24 08:20:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ce92c1f593fb30d367e59e514d05f790e1d71f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 08:31:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/41/672441/1/check/build-openstack-puppet-tarball/689f868/ : SUCCESS in 3m 16s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/41/672441/1/check/puppet-openstack-lint-ubuntu-bionic/f53e073/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"705a0a1e9cc4c2ee4db8e0ec7adfff02f877149d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-24 13:32:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72dc5036e60784a7e65326223ee8519dc5759d0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:32:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85dc15b0d7b8e54b35a7fa374bc90bb3e6bfaab6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:45:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c2480a224d3cb68fc3bcae3c820330d304d83e06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:45:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/41/672441/1/gate/build-openstack-puppet-tarball/9fe7c22/ : SUCCESS in 2m 25s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/41/672441/1/gate/puppet-openstack-lint-ubuntu-bionic/557340e/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4525bc4c0d843974540b5b8c6bbe3ee6575b1185","revisions":{"4525bc4c0d843974540b5b8c6bbe3ee6575b1185":{"kind":"REWORK","_number":1,"created":"2019-07-24 08:20:27.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/41/672441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/41/672441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/41/672441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/41/672441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/41/672441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/41/672441/1"}}},"commit":{"parents":[{"commit":"08476e4e737ecf93dac4529071090738885e4bf2","subject":"Change memcache_socket_timeout to a float","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/08476e4e737ecf93dac4529071090738885e4bf2"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:38.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:38.000000000","tz":480},"subject":"Prepare Train M3","message":"Prepare Train M3\n\nUpdate the version metadata for Train milestone 3\n\nChange-Id: Id1ae971d8f099b770841db382269f5ed039d8069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4525bc4c0d843974540b5b8c6bbe3ee6575b1185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4525bc4c0d843974540b5b8c6bbe3ee6575b1185"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
