)]}'
{"id":"openstack%2Fpuppet-oslo~715815","triplet_id":"openstack%2Fpuppet-oslo~master~Id6a016a2b3a46f231d77d7f676d4a42c520d283e","project":"openstack/puppet-oslo","branch":"master","hashtags":[],"change_id":"Id6a016a2b3a46f231d77d7f676d4a42c520d283e","subject":"Add CentOS 8 and Puppet 6 in metadata","status":"MERGED","created":"2020-03-30 07:32:22.000000000","updated":"2020-03-31 15:49:58.000000000","submitted":"2020-03-31 15:49:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715815-1585669798076-eecf9550","meta_rev_id":"10a3b0312b58b69b2d39774f5475c018c755297f","_number":715815,"virtual_id_number":715815,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 15:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-30 13:54:06.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-31 15:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-31 15:41:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:54:06.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-31 15:41:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-31 15:49:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3485d11fc6256839ddbbbb8006c9766e32463700","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:32:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18f9203e028c43351c85692a0310419f6d9b8e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 07:42:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/bcfaa04dabed4649b73fbb684851b540 : SUCCESS in 4m 37s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb071fd0a1f84d0c8251af4aa7f242ef : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"25ab1ba0c159ba69bc0aa59d244baf8f2a800cfa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:50:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"16af2109753471887a0dc6f4725d53c1211f784f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 08:00:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/9f07f8400dac480988352888d73c3e03 : SUCCESS in 2m 45s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7543fe53d16b4106b9c37e06a4fe0557 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b4cef4947f8811240f0f89c0cf4d812f5d50ca01","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-30 13:54:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62ec36c95668d8b77543e60edfd7c8f746fc7841","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-31 15:41:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f307b84b90a75482f72cf7d7bab37e0e97ba8e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:41:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"945dbea69253e1469d37bae26b260f7c1d8a0fef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:49:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/e2baddfbe66742528a86f13c3cc5d13e : SUCCESS in 2m 55s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/497c17e3fc814ca3ac72c10299c25f95 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"10a3b0312b58b69b2d39774f5475c018c755297f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:49:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e947083aa6c28350e3c7711012e2e37466baa87","revisions":{"42f287d60e2073c1da64ad29c05d384eec15cedb":{"kind":"REWORK","_number":1,"created":"2020-03-30 07:32:22.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/15/715815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/15/715815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/1"}}},"commit":{"parents":[{"commit":"1338b26b02a76735e53968eea1ea28a2cb68a1af","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1338b26b02a76735e53968eea1ea28a2cb68a1af"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:32:15.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:32:15.000000000","tz":120},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: Id6a016a2b3a46f231d77d7f676d4a42c520d283e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/42f287d60e2073c1da64ad29c05d384eec15cedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/42f287d60e2073c1da64ad29c05d384eec15cedb"}]},"branch":"refs/heads/master"},"0e947083aa6c28350e3c7711012e2e37466baa87":{"kind":"REWORK","_number":2,"created":"2020-03-30 07:50:30.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/15/715815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/15/715815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/15/715815/2"}}},"commit":{"parents":[{"commit":"1338b26b02a76735e53968eea1ea28a2cb68a1af","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1338b26b02a76735e53968eea1ea28a2cb68a1af"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:32:15.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:50:29.000000000","tz":0},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: Id6a016a2b3a46f231d77d7f676d4a42c520d283e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0e947083aa6c28350e3c7711012e2e37466baa87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0e947083aa6c28350e3c7711012e2e37466baa87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
