)]}'
{"id":"openstack%2Fpuppet-oslo~724007","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fussuri~If9ce0bd172ae0b75ee678d0fd178ab7082f5de8f","project":"openstack/puppet-oslo","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"If9ce0bd172ae0b75ee678d0fd178ab7082f5de8f","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-28 14:57:00.000000000","updated":"2020-05-04 07:01:26.000000000","submitted":"2020-05-04 07:01:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724007-1588575686459-d77804d9","meta_rev_id":"09c0671f7481c177b098b97d41da830f8191a0f3","_number":724007,"virtual_id_number":724007,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-04 01:40:58.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 07:01:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-29 15:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-28 16:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-04-29 15:22:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:47:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 15:22:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-04 01:40:58.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-04 07:01:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce66975dd13ce5f74f071c7e3839e6d14c22d8e5","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"825e42e148ff12ba00af1eeb789857c4ece272d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:36:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c0aa81ab3194950b5edebf7b19ee768 : SUCCESS in 6m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/145de427be254680b8db1802c765d458 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbde4e7dbfe84270b21e69745a58d805 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/572ba3f006374c548431e5273085a168 : SUCCESS in 13m 32s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a55b9cfe082548439095780283bb6468 : SUCCESS in 13m 30s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1a95cdb5157446abd31dfec0aab384f : SUCCESS in 10m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b00a0686b5f54986b4349ba99a6addb7 : SUCCESS in 11m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3ea7de08daef47d282cd5cf176601d53 : SUCCESS in 1h 10m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db446348054e41ecaa9eb5400969fbc3 : SUCCESS in 52m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d6d4a89a51a4638a979b589b6ddb90a : SUCCESS in 49m 42s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4a473de523454ff9a7b24bda0e85647f : SUCCESS in 45m 16s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/db4b449282834c108c6478b444689e13 : SUCCESS in 1h 04m 03s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8319ebc563504b7688660d73dbc0e174 : FAILURE in 1h 08m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c494bd6315004cb786da29485f986018 : SUCCESS in 1h 10m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bb24125b64104786bdea688941fb2cd2 : SUCCESS in 50m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/78b5422436f24537991cf276bc529e3f : SUCCESS in 10m 39s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e7b335a1e28f4c0682a66a9292e722a4 : SUCCESS in 12m 18s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b2198502b3745c1bb67fbfaddf5b200 : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d768611e4f0c81fa031887f6e713e74685a64670","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:47:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2b42914d4992ee142eb93b12d860ae1c67e8385","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 15:22:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d368e2660ccd9c1671e0c555be004093684467a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:46:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/014850871721446ab440351da9f2faad : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/971678440e794f5c855fe2ea6e6931ff : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2203c7721e62472f9212f44d48a5fb4d : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/33671ca3f523457c8ab248718f9b9713 : SUCCESS in 11m 47s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8917c86a842d496995b308863eb378eb : SUCCESS in 13m 49s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea096c5d6d4e4f94b4b091b288680787 : SUCCESS in 9m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fc95991ab014ae0b4a6c8dd8782ebf5 : SUCCESS in 8m 27s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dfcd39b6df2a4bb8b80a87901ab2389c : SUCCESS in 44m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3afc430c523f4c4eb38f9bcc2bd31cbc : SUCCESS in 51m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff95bc02bcef444cbb6fe7314c0621d0 : SUCCESS in 42m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/27d07a89cb68497eb4ad9f7e8c5166ed : SUCCESS in 34m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/115fac1c35a342d38ba8004736be9ec6 : SUCCESS in 54m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/57a042f7e0f54e8cb8ef5346731664f5 : FAILURE in 47m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f09a3763ea449b9972d05574a696ee3 : SUCCESS in 1h 19m 30s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4e0e8831c1e74b909ff68ee5900fc97d : SUCCESS in 58m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/12c1f907af2c4393af5c4e9e074f33ae : SUCCESS in 10m 17s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d41703edec1341e0bab1a68a31488bd6 : SUCCESS in 9m 13s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edf56306b1274a4d8406db9d77c161ba : SUCCESS in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0079d34622077fb8baf315791983afed1959a23b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 16:53:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec44f167e651797636fd31c45bcc6840ee96cd6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 18:05:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c9e614f1d3846f4a8c43d5bc583456c : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c5e58a7a0694742af6164bd1b5f47aa : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94c8be45e6d74078a0e533bf98f96588 : SUCCESS in 4m 46s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e8b41558383a4ba495efd3ec03a087ea : SUCCESS in 11m 49s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/e15fa682bdae43259103283ee714716d : SUCCESS in 9m 35s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ab54014bb114ad590586cc85a803e54 : SUCCESS in 9m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc412c115c9940a396bdf04dd893aa4d : SUCCESS in 13m 16s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/08eca8abfc694a65904dbaf05c0f5353 : SUCCESS in 1h 00m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77dfe319e41d403db886732241c600ff : SUCCESS in 55m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3d0387a103e4d169d51fa7935da0b02 : SUCCESS in 43m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c40db6b1d9534675a4bf5274361c3ba4 : SUCCESS in 34m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/159c45ab2f6747388681eedb8eaa0278 : SUCCESS in 59m 41s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6328348aef4e4c789c57ac44da76bef9 : FAILURE in 40m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2fe942fc698340b9beadf2246eb3d29d : SUCCESS in 1h 06m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/11eab315a43a4949a7040247e9b5e500 : SUCCESS in 44m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6b6e47afc83746c99f45109a93266906 : SUCCESS in 9m 32s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a6dea718f19342b2a224ab97593118e4 : SUCCESS in 10m 23s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd7e697b58b14479b86bbca2e433d638 : SUCCESS in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06eed151ee0925be97963cda9ba6df2c634aad89","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-30 09:56:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9bf4c70d38883e6c0829e7f6cf71e28d6a7435c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:25:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35dbcae3029840c6bf070bc3f060d17c : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd2bd98c30b94e43a0421b8acb281e95 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64fcd2812a5a47ed9dad6b90e24aadc8 : SUCCESS in 4m 42s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e0ecfb7f336f4dbebbd9100ce09e64de : SUCCESS in 11m 09s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/edb373e6579341d48615813d8a29bd14 : SUCCESS in 10m 25s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b02752895b7e4e999eb35295f2278256 : SUCCESS in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/804c5be3de914d3f84c77a6d813cd242 : SUCCESS in 10m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9cb5c4f28fb5476fb84c662206225d42 : SUCCESS in 46m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fdc425dd8ff417da8dc9b14a1edc9de : SUCCESS in 58m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59c968f8305e43b3b643486e7bf74e1c : SUCCESS in 39m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b9be6c49390d4ad08829394a23e0b59e : SUCCESS in 32m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/44856df571e944e3817899166c6287c2 : SUCCESS in 1h 00m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e4c37c606004d1286b93760a833d9aa : FAILURE in 1h 27m 31s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f95fd590f51d41aca756ea95eeb86e65 : SUCCESS in 1h 12m 35s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/18986c3158084a099382bbb87ae17f5b : SUCCESS in 1h 17m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9f4d4aecefac4bb987b2b3e696344bb4 : SUCCESS in 8m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4b561416e7074f47a64945f931f599e7 : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5eddc1aa727447f94bc540b1ea330bc : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bc9f8c24142134920b10415cdbc4bc3e8e158e8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-04 01:40:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d1b0744f75ed5f437f78850ccf1776bc1ab76b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 03:38:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/520c13b1fdf54b07b6743eff0f853aae : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fae2dfaa825455f902d1d328216b6a3 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ccc83cff5bb44e7904ffa0b42b65ed6 : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2012f4696d654455add5a5f5402c672c : SUCCESS in 10m 49s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/41592d9e9d8442d5b01dedd9a6778366 : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d918b9ceb6d48d6a77e905fe62970a4 : SUCCESS in 6m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c41662033ada4f5f9d0eadc5804d0c81 : SUCCESS in 7m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dd57f8163e3c48598fc3a1301c100153 : SUCCESS in 44m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e484848389ae4eb88bbbc835d38ebc75 : SUCCESS in 51m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29d6d5f9a65d41e79492a71e78fbdc7f : SUCCESS in 41m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a9a84fdde0e342509dab01ffeec583a2 : SUCCESS in 39m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1d4b83cb8fb64720a97b52446a811bb6 : FAILURE in 1h 56m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/39f82e31b1f94324b923a117aecc689c : FAILURE in 58m 33s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/56e59d14b6494cdcb6d803c72b710e31 : SUCCESS in 1h 33m 09s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/15b3776b10ab4779a09a3eef8dc332f9 : SUCCESS in 42m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/feb748ce63f54f68af43d48e7470d7a0 : SUCCESS in 8m 34s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b7a918bdc7934c069a4fddb13712378f : SUCCESS in 7m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/214cdaa3ca97493199a84cd7b6f20437 : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f503b3921ca1d96ff1438fafa98eef9d7f02a18a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-04 04:38:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"edd873a021e946cb0732917ef0bb3332270db515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 05:43:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f2c899c62c24b04b7a57d76f0ded448 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4d34f5e94e94efebf90117f04fc7c84 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69725bcf3d9d4b5e94f0ef1e33d711b6 : SUCCESS in 4m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/63860a4d17714e539fa99caa805406db : SUCCESS in 11m 55s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/928c9efb3eef42f4beb36b124f8e1b55 : SUCCESS in 12m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e93b0edeba041eca1979474b96bdea8 : SUCCESS in 7m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b29cce0033f34800be702cd4939fb738 : SUCCESS in 7m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/94ed6a5a5d7247c78560775545e5759b : SUCCESS in 44m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/448266af8815487386ede19146864a80 : SUCCESS in 49m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6eac99b6d054b1eb67bd2c2fbe2b20f : SUCCESS in 41m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6b8fdf704f6a4bd7b73d52168d1efc8a : SUCCESS in 33m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/399ae1c1bf044a47a0e3df127b8a1936 : SUCCESS in 50m 38s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c64b8d3f75e14a518079b4800cc70a5a : SUCCESS in 1h 03m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aae10a4c4b874cf3bfde57f045ef1399 : SUCCESS in 56m 57s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b90f57f2fee847619d8addb62e218a61 : SUCCESS in 44m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4252305897a349f08265882812a3bc2d : SUCCESS in 8m 35s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/777351b0f2314e5a897295f34407d59f : SUCCESS in 7m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c785a0317e94615b8e998bf67155605 : SUCCESS in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df9c0aeebce4b14a6b9b381840ec72ea352cdb58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 05:43:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfed3169985fa91a2b6b2d7c509f0b649f7ec6ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 07:01:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1cca4e7444f7167a43fb171d229a361747512779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 07:01:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1086e0c724b45f2a6fefcec8923bcff : SUCCESS in 5m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21cfd1cf43e147608867bf1fb2c55311 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89fba8225b3b47dc91f7548610021354 : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7c9dbc794b5044a7bb6d06dbbea20779 : SUCCESS in 11m 11s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/9d6b3a1a2b01481b82e9a85ba663731c : SUCCESS in 14m 16s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e6bd403f40a40928f2aa994558559d1 : SUCCESS in 9m 04s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c7b665868b82409785b8e21d88a8a581 : SUCCESS in 54m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a22d11c2667b4fb0a312ed446cc76842 : SUCCESS in 1h 05m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c58f6e5709594f3f9e71bbc169b636f9 : SUCCESS in 1h 16m 23s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4d56dc25eef74f318dd78b3cd1820fae : SUCCESS in 43m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0491f3b659bd4b0dba7b7cbb8ea0ff02 : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3251589a98b12807cadfdddd0943bdbb81c1cb50","revisions":{"3251589a98b12807cadfdddd0943bdbb81c1cb50":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:57:00.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/07/724007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/07/724007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/07/724007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/07/724007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/07/724007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/07/724007/1"}}},"commit":{"parents":[{"commit":"62ab58fe78d5f32b81886fb4c4084097eb9e307b","subject":"Prepare Ussuri RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/62ab58fe78d5f32b81886fb4c4084097eb9e307b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:56:58.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:56:58.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: If9ce0bd172ae0b75ee678d0fd178ab7082f5de8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3251589a98b12807cadfdddd0943bdbb81c1cb50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3251589a98b12807cadfdddd0943bdbb81c1cb50"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
