)]}'
{"id":"openstack%2Fpuppet-oslo~724008","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fussuri~I0991f7b2fbe13753740e5aa26b39ed7ba42618dd","project":"openstack/puppet-oslo","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I0991f7b2fbe13753740e5aa26b39ed7ba42618dd","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-28 14:57:02.000000000","updated":"2020-05-04 08:30:04.000000000","submitted":"2020-05-04 08:28:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724008-1588580921547-62275d76","meta_rev_id":"f1d1b00e40f6ed27a7a3dc8ee0187039fc422216","_number":724008,"virtual_id_number":724008,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 08:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 01:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-28 16:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 01:53:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:47:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 01:53:04.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-04 08:28:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9e0ad8c76db302be30975c2200fa51fd1b5b91c","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:57:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4b2317d5e30de0c27c423bc56563b613bdf0b7c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:47:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"185ef893c091c98e314ca1c0c762f6d17fcf8ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:26:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3c237ff564549adbd3d913aa8697de9 : SUCCESS in 7m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/203ab7685f6949269fd440456dff363d : SUCCESS in 6m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecb4745854d14ba6802c4b74ea178081 : SUCCESS in 7m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/87cdab9ca108406db173ec0a7de2c708 : SUCCESS in 16m 48s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2794e9c09e0a48d9a189c3fefb33b22b : SUCCESS in 13m 13s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fe94c197df045dc838d5732404b85b8 : SUCCESS in 8m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/662b699d758b4a8f82d91575200ca01b : SUCCESS in 11m 58s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d2c717a6d9694fb3a7e5161573e55ddd : SUCCESS in 48m 42s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f6b358318494d53ae7b8456312cb208 : SUCCESS in 53m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e12cf198d5a446d2af231e5ac7a63440 : SUCCESS in 43m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cf68b676b25246afb39f00e2d3c93343 : SUCCESS in 38m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9f75f5cc54044106b83f36d2ee362977 : SUCCESS in 54m 22s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/09cd47f3914043f8a8a4ff5c9c72a48c : SUCCESS in 1h 06m 34s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6c40464227f5434e8b0faf926b3ba8d2 : SUCCESS in 1h 10m 05s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9ff75e7288a4862ae84e19661585dc5 : SUCCESS in 50m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/972708a27fa6420c845bf02688e409b1 : SUCCESS in 11m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d2ecd56a377b4c02b24ea3b5fe6ef8b4 : SUCCESS in 10m 49s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37bb40ca47404ad98b038141fe4ebe47 : SUCCESS in 8m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac541561c5de43138591b48ba615af9c : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d9cc9901bc24839a17ba31c660d33da05c04ee22","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 01:53:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3531e515d6333cef1d9fec708c23109eedabcd16","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-04 00:13:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67d0ad781b44eeb221d85600ca415af8dc0307c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 01:20:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81ab8c6e6e094c99b8ed440c868e5cbb : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26d03623aee64dc0accc7f0effc18381 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ef3f32919804242bd028731d4604c3e : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f46d3e0ba5dc476bac503253cccca88d : SUCCESS in 12m 33s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/463c9d9b9c624321989d65c8ee28f06a : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73313f1ebf50471c96b23527afdb0591 : SUCCESS in 7m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94f0f78429934eff956696676a849542 : SUCCESS in 11m 25s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1af861d2374e4553bbbfda95d7e40ffc : SUCCESS in 49m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dbb36b39af24bc7afd1cf8dee06c9fa : SUCCESS in 48m 16s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14392c26d7944f058a8bfce6bd1cad15 : SUCCESS in 44m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/79d4a7b095c849d3b1cf2e55f40b0aa6 : SUCCESS in 33m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3fc0aad3df16439aafe28f82941b8faf : SUCCESS in 53m 16s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/087c113fd3c14e3bbf814db3cd629ba5 : SUCCESS in 1h 01m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/298a7597a506429db5b8668a2d4e66ed : SUCCESS in 55m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9c1802f949a04bb48a7c8f52146b9f2d : SUCCESS in 46m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e8749a31115d4862a905836daa4915cc : SUCCESS in 8m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/63519f70c4bd470e9489f063b121dce7 : SUCCESS in 10m 03s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7eeb46587b904fff857900f20dbd0c4a : SUCCESS in 6m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/743ce651367a43048c7bac21dfba2dde : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2bab0b25c58aafc49d8c6049a5f25fee0504447a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 05:43:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9794ec95d4c163a4aa4da398c1734548254c2223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:28:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f86e1af004645f0b03b2940b42413e7 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/376c05c0ad5e4b95be16178329b7bb03 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7940a3ee24c243a38c1065f5e405c963 : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9d232fe8402e47f7858a1a0d193361d8 : SUCCESS in 14m 19s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/efc58efb4d7042a0b4b7f501d31cd832 : SUCCESS in 12m 24s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6db77f95ac444f28a12d3691b4c779e : SUCCESS in 6m 53s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/04502e5849904a259680b18c354ac725 : SUCCESS in 1h 05m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f852de537f304375ba3fbd39189ae90f : SUCCESS in 1h 05m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/569309ea5ca242bbaf2249b5713e8239 : SUCCESS in 1h 14m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/72c21498798042c4954bb26acf2c2e94 : SUCCESS in 47m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/818ca5c2bd9548d8a8ed55d6c73a6070 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/701c763e7270472ea0db1edbdea6e3f3 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1afcdda8db3b57329fc77a0ff9883f1dd7d85962","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:28:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e41533ef15d7a53179cc392ecbe98fcff409ca61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:30:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d77018328a44646b1bc0e176f144eca : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"602de85419d8a502a95751ca54a4359c6ecab0bd","revisions":{"602de85419d8a502a95751ca54a4359c6ecab0bd":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:57:02.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/08/724008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/08/724008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/08/724008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/08/724008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/08/724008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/08/724008/1"}}},"commit":{"parents":[{"commit":"3251589a98b12807cadfdddd0943bdbb81c1cb50","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3251589a98b12807cadfdddd0943bdbb81c1cb50"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:57:00.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:57:00.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I0991f7b2fbe13753740e5aa26b39ed7ba42618dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/602de85419d8a502a95751ca54a4359c6ecab0bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/602de85419d8a502a95751ca54a4359c6ecab0bd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
