)]}'
{"id":"openstack%2Fpuppet-oslo~725166","triplet_id":"openstack%2Fpuppet-oslo~master~Ib327d97b13877b5ce2651e5a4f15b9d11c939000","project":"openstack/puppet-oslo","branch":"master","topic":"oslo_config","hashtags":[],"change_id":"Ib327d97b13877b5ce2651e5a4f15b9d11c939000","subject":"Deprecate oslo_config resource","status":"MERGED","created":"2020-05-04 05:40:09.000000000","updated":"2020-05-04 18:56:34.000000000","submitted":"2020-05-04 18:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725166-1588618509896-ae45bbcb","meta_rev_id":"5213dc77c839ec98de8c80342572d509ea0e2b25","_number":725166,"virtual_id_number":725166,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-05-04 15:46:14.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 18:55:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-04 13:02:58.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-04 07:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-04 13:02:58.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 07:02:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-04 13:02:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-04 18:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab119519ad1bef2e2806a85eac9d4814225755dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-04 05:40:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc2844144e4188818f07311c9973f1b4a29c00da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 06:54:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91c647dda3514ea2890a454ca54f22fe : SUCCESS in 5m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54dd4a85b0ff4eb8a737e33d38e5ebd8 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b57b03b35b7248089d3ff0047725501e : SUCCESS in 5m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2151b11efe4a44ae852fcec346c48d7f : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c443f7a1db4349c388e78f3b6188abfb : SUCCESS in 14m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01e9d21391f44da2ae37c5d8409f758f : SUCCESS in 9m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17f8ff2e1a7f48db8de5838c52024b34 : SUCCESS in 11m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f3425926b1fc4893b72b321e5ab9379d : SUCCESS in 59m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4719603752c4e59af83da754a9d7e65 : SUCCESS in 48m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84221ec4590c4578b8a85ad9d5bfc9fb : SUCCESS in 42m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/227f05a92fc34c0c9327c78d8c4a74d1 : SUCCESS in 34m 35s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5b6abaa700fa46c18a4430d3af24f0b6 : SUCCESS in 58m 40s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7d81b3141018436eafb21ee7169b69d5 : SUCCESS in 1h 02m 24s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bb67dc2901ee4c9a96f48860b535d1ff : SUCCESS in 1h 09m 38s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/867954486b2c448aa2eb0fa9652254c7 : SUCCESS in 50m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a0364302cdf44d8bb77927fc8dec5d6a : SUCCESS in 8m 10s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b749fb6ad48b485ead6562dd5801d109 : SUCCESS in 11m 56s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/848caa569c4b4335822e63eedc399d40 : SUCCESS in 5m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cfb48bde6df4131bc7bd38f7a7817d2 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5158c6bfd6136899f57ab8c35e673f5ae0a7e32e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-04 07:02:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da178a97b5c13bdc3a73851a875c41c5f13eae96","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-04 13:02:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05745e943076a345ad36a6524e269b135692b7dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:03:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4594f18ff07d8ad8148121ccbe486c88bbac6f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:25:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8c578fe2719474082be38adf2040cff : SUCCESS in 6m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a54f727908d4bbfbdeeb79249d4d421 : SUCCESS in 5m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bdec4f3329a44ca81bb6f4c93b5e23f : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/80d3bcb411d54a6c8cfe5a20abedd13c : SUCCESS in 16m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e418c3af1f59436a95d282595d559046 : SUCCESS in 14m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b89b80bf91e4acd8b17260aa031f51d : SUCCESS in 10m 12s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/283c028dbf304964ac82399906abd136 : SUCCESS in 1h 01m 40s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b668876205744c98505df28fde0fcd2 : SUCCESS in 1h 16m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/61805a00a3774a7d97609835f5902499 : FAILURE in 25m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eb4118cb01e34c7a8b50af66c123a4dc : SUCCESS in 42m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5c4b6f0fa6bb452a8596e6cd62e1d5f1 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb782fe80dde4c4fad3def62fb4bfd00 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5d0740427c4ad9968aa85e4a8fb493abdccd9ee0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-04 15:46:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"419e1a711b20dba38aae367fd063ff8c2784a84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 17:17:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d344068d36b4857935674a0ebd4884f : SUCCESS in 4m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f021f41f27b47988682b4ee1b02bbee : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cff561cc269442af8267533496cb006e : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7e8dfb68f2184df39f39a24c71ee6793 : SUCCESS in 13m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0e29c30b3420474394e1d9e7863773c6 : SUCCESS in 9m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96e946664a02405ca0478313cba55612 : SUCCESS in 8m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d6f69ced30f4c9db755f47301331d03 : SUCCESS in 8m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/65c0783ecdcf4a878b6810c129b4aae9 : SUCCESS in 1h 01m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba631ddeb6a140f8acf4a55711eaa339 : SUCCESS in 49m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/530dc2628a1c43dcab37634df3f46c55 : SUCCESS in 41m 33s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/63d5060d11ad4df985eacfae77fdf467 : SUCCESS in 33m 07s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eca8abd806d44497b68b0022d182881f : SUCCESS in 1h 00m 56s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/29abee3145d54039b4d39caf519d59bf : SUCCESS in 1h 22m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3019479d1e7a4e32b6de9177c8120c76 : SUCCESS in 1h 12m 26s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/de1cb1be642f4a6db10584074ea5ae26 : SUCCESS in 49m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5065fca0bd0e4aa4b420c9c7f46b7f34 : SUCCESS in 11m 25s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ee4344b034d040f69dd4bd546502a662 : SUCCESS in 8m 33s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c5940957da54d0e816aef928fb5eca5 : SUCCESS in 6m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db2d29e2291240618ceb102482b1f334 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c9815036c117faced48e873c10ef018797b28599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 17:17:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c356563d17c8af57bca6093ea349d8bc6bfc9fb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:55:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"753d3f9a6525639a8eebd713d9c43e3a1f9c52ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:55:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f640b493fb74ac3b46eab3ea902696f : SUCCESS in 4m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f1ecc81c80d42c192601d060ffdf110 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01e4515d97b14906954bb3ef33f1d22c : SUCCESS in 4m 59s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d69a02f787cf42d985e7369d063bc0de : SUCCESS in 15m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3ea90459b923414eb405e546191b8143 : SUCCESS in 17m 50s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1af9dc357cbb4992a5acd697b70477ef : SUCCESS in 8m 49s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7dbb1bfa487a4b3db84a6e4d4adeda56 : SUCCESS in 55m 04s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0c17a74d54ee42e9b3effb3bb0d8e52b : SUCCESS in 1h 20m 46s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/49cda31d7f0049e5a9ebb499b1495a3e : SUCCESS in 1h 26m 32s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8587b3a8aa2e463b98f2f93a5d0630fa : SUCCESS in 45m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3ff41ad460dc467a9af29e3355293efc : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a52f958f417e4ce2b33756657ea4856d : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"88756599184d38d25fe48ee41028494de10b4ede","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:56:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2d4c522b4c149ceb0ebfc043d4362bc : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6cd2bc408d55c7bdaf43013e1a2c0a1c4fd1fa9e","revisions":{"6cd2bc408d55c7bdaf43013e1a2c0a1c4fd1fa9e":{"kind":"REWORK","_number":1,"created":"2020-05-04 05:40:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/725166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/66/725166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/725166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/725166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/66/725166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/66/725166/1"}}},"commit":{"parents":[{"commit":"b8d2c4c326ff0e5ad7a7cee915b7dadedff70214","subject":"Runs on Debian 10.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b8d2c4c326ff0e5ad7a7cee915b7dadedff70214"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-04 05:39:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-04 05:39:35.000000000","tz":540},"subject":"Deprecate oslo_config resource","message":"Deprecate oslo_config resource\n\n... because /etc/oslo/oslo.conf has never been actually used.\n\nChange-Id: Ib327d97b13877b5ce2651e5a4f15b9d11c939000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6cd2bc408d55c7bdaf43013e1a2c0a1c4fd1fa9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6cd2bc408d55c7bdaf43013e1a2c0a1c4fd1fa9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
