)]}'
{"id":"openstack%2Fpuppet-oslo~729581","triplet_id":"openstack%2Fpuppet-oslo~master~Ib66999db407f6b8872196f105e49a6296e42528a","project":"openstack/puppet-oslo","branch":"master","topic":"zmq-removal","hashtags":[],"change_id":"Ib66999db407f6b8872196f105e49a6296e42528a","subject":"Remove support for ZeroMQ configuration","status":"MERGED","created":"2020-05-20 12:34:43.000000000","updated":"2020-05-21 18:11:49.000000000","submitted":"2020-05-21 18:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729581-1590084622493-4cad9dd9","meta_rev_id":"882f48355c1002709f8617c8d90f18b3df007d9e","_number":729581,"virtual_id_number":729581,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 18:10:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-21 13:37:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-21 16:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-05-21 16:48:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 13:37:47.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-21 16:48:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-21 18:10:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"714c5c95ccb2f98dce53ff1e088f65f1e7927af6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-20 12:34:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"686ecc53d48281882a48ff293647a7f98b053ad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 13:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b12efc90c4db4338b642888e32c8d2b8 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6305c2ba2c2d4ad4baddd286d3288dae : SUCCESS in 5m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad7d2907c05247cba8701b29b722d4f9 : SUCCESS in 6m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/67cb51b616884646b80ffcb3a4add0b8 : SUCCESS in 15m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d8a103e3ef634dc8bab68fa58a1699a1 : SUCCESS in 16m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bca071adc464c669a6dbcc448cbd031 : SUCCESS in 12m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0818fb08d3349b5bf431af3f5f05a22 : SUCCESS in 12m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8b38a32729c44052a14e949e5a16ea5c : FAILURE in 39m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28bb21acbff74a25914deda851f41444 : FAILURE in 41m 29s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/816d2882877241e2a87e64ca7e999011 : FAILURE in 35m 48s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c0890b2025c4446ca5066c91f9500075 : FAILURE in 30m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf473dd621934a1ab6e60a343587c150 : SUCCESS in 53m 29s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/867adc9e8f674ae0b0fff9e128f9887e : SUCCESS in 1h 03m 48s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/511e628d791345d58fe088615260af62 : SUCCESS in 1h 09m 17s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9a989b1d9324855b617dc2858c41aaa : SUCCESS in 42m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a00e0e168deb49409a5ae558a7d505bd : SUCCESS in 11m 08s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b9a074f6a5204e06b6c44a556c70c99a : SUCCESS in 11m 57s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f16b192cf4b4e6ba3f39d0ab18c064e : SUCCESS in 9m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcee6b088d3d43c5a0af3de2f6daa6ff : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0acd785840f0b3ea39f80b85f1166b465269a78f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-21 13:37:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edc8140615ba17eb905156a4df871af9bffa9d49","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-21 16:48:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5bdc3648a23e3628a011ab592de141908b30bec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:51:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f459cf4b845f6f915505371967f96ab0ce22169","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:10:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3027a9464432ab996525c6c8b8e067640d915fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:10:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23f9da347ff04ab9ab054e16d5339b07 : SUCCESS in 6m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/822c7dafef5c435f94be49b3b559790b : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eeb560a8dfca4a1584a1293cff99ccc3 : SUCCESS in 6m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0b3f764b04574ba1bf6edbd06da35fe2 : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6ee43b4f23734afeba112613d2e8daf9 : SUCCESS in 18m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0f8e8a13a0c4b5093057009c5e1d758 : SUCCESS in 12m 43s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a232199668cc4ef4857334610ae5f276 : SUCCESS in 1h 13m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f6bac3ce1c8454f82c8d0674633e29c : SUCCESS in 1h 04m 13s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6e960e7a0e1c4d93b95dc61af8b4c6ea : SUCCESS in 1h 05m 05s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/10be130df1e34c3c80a593ef49e51067 : SUCCESS in 47m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/be92c415104c4f5b8f2dc122c394ecea : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7da0e714e76a42b8bb5b2ad334d8d264 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"678abf93125e8b6ee36323c9b3c96317d234bcc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:11:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a88d1f17c7d457dbbb27763346cfd58 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67912587fc76fe4633883002826dc2c8ae6a839b","revisions":{"67912587fc76fe4633883002826dc2c8ae6a839b":{"kind":"REWORK","_number":1,"created":"2020-05-20 12:34:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/729581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/81/729581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/81/729581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/81/729581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/81/729581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/81/729581/1"}}},"commit":{"parents":[{"commit":"c6fb0f815d77ce3a31965187c235795f8453759f","subject":"Bump version for the start of Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c6fb0f815d77ce3a31965187c235795f8453759f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-20 12:33:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-20 12:33:30.000000000","tz":540},"subject":"Remove support for ZeroMQ configuration","message":"Remove support for ZeroMQ configuration\n\n... because it was deprecated long ago and ZeroMQ driver was already\nremoved from oslo.messaging[1].\n\n[1] f2fd8d437589d94443a137c7ca7a38d6b782d153\n\nChange-Id: Ib66999db407f6b8872196f105e49a6296e42528a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/67912587fc76fe4633883002826dc2c8ae6a839b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/67912587fc76fe4633883002826dc2c8ae6a839b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
