)]}'
{"id":"openstack%2Fpuppet-oslo~746248","triplet_id":"openstack%2Fpuppet-oslo~master~I8ff8886b65a323a65dd0ff9d01066c82a4116f8d","project":"openstack/puppet-oslo","branch":"master","topic":"trivial","hashtags":[],"change_id":"I8ff8886b65a323a65dd0ff9d01066c82a4116f8d","subject":"Remove release note file in wrong path","status":"MERGED","created":"2020-08-14 00:22:20.000000000","updated":"2020-08-18 14:34:44.000000000","submitted":"2020-08-17 17:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746248-1597684131156-7628fdb8","meta_rev_id":"02a8ff7cc2d110678c06f94ab92f629e008e4deb","_number":746248,"virtual_id_number":746248,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 17:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-17 15:21:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-14 09:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-17 15:21:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 09:05:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-17 15:21:37.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-17 17:08:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"879ceee444fe63c0b9cfeb58a292a0ee37d97e1b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-14 00:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"829c217fe586d3abd841ef2a168b5bd8bacedb9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 02:04:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2da5759172964defa8f50d183523bb94 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37a8860ad96b4f08b7877821669f66bf : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22739570ae154edb83226ff12634daa5 : SUCCESS in 4m 44s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/dc65e8a5990f420592df6d66b02d60c2 : SUCCESS in 12m 07s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b9a6fa8778894ecc8b022dfd991e7401 : SUCCESS in 12m 47s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5262edc86f5f4da481d7bb72a07bd253 : SUCCESS in 9m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40a20c4925ba431fb5ee5f0f25a4638e : FAILURE in 7m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3d96d7dbece84d3facc6a648a7787b14 : SUCCESS in 1h 11m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/934ba32993db45aeb33c50ec41d9ef34 : FAILURE in 1h 07m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cb0569823324f38aa318d5a68dbf3f6 : SUCCESS in 1h 07m 48s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f721d0c8e17e4923b52cc7747da64a77 : SUCCESS in 52m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8b0cecdf9da5412d887b646d92ca03f5 : SUCCESS in 1h 24m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f845ea0c654446e581aa0dfe33669c93 : SUCCESS in 1h 32m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/46174bff6907417baa2a3404c83251a0 : SUCCESS in 1h 39m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4266520679f845e6bb7da3976480b6dd : SUCCESS in 54m 05s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d766fbebb0d049f8ba45d9201167fed2 : SUCCESS in 12m 07s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f18c43b641ae44ecb6c1a415aa10a42b : SUCCESS in 9m 51s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b879162819394f358dd78e5ee2652b38 : SUCCESS in 7m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"194282de3d4d07d7a0333a3cb30431e1c0a74079","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-14 02:19:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5740cb47d66a757f65e294740f87d0d720502c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 04:00:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b991ca00ef7434ba28bb802f2ec7842 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1abb9cbf911842a69f213f18fb7f0925 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71dbb269f5bf4535b7016533c9bc8b91 : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1281b6115e8149fb87d04c71ac59b3a8 : SUCCESS in 15m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/984415b3c05e4afd9fc861938d01959c : SUCCESS in 17m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af460d7791514094b7b1026608bb6952 : SUCCESS in 13m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33d932d996594affb8eb4cb2be90b987 : FAILURE in 7m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f0f9e42ed103481e8aa626529d393278 : SUCCESS in 48m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c05db54e77004a20bcdf5afb4cff3f0f : FAILURE in 1h 03m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc541cddd559456f9e4608acc80424c3 : SUCCESS in 57m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5da14ea0730f43268f18bc8698a4cc40 : SUCCESS in 37m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42ff7214945c47a8a67ec46993d876d2 : SUCCESS in 1h 16m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4580cf7866ea4b70a99832d19745e5a0 : SUCCESS in 1h 21m 05s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/088d9c6932f14eff94efc3dc938167a7 : SUCCESS in 1h 39m 05s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6a36f89b7f0d4d4ba347f4416fc5ea50 : SUCCESS in 55m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4c3e0b05d6ff451eba5e5002c77a7ed8 : SUCCESS in 9m 54s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8e3d74f72de14ee0884d03b88f66f46f : SUCCESS in 10m 25s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95a3c03bb3dc4890baefac1d9ca25adf : SUCCESS in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38dc4e1e50d81133aa195722370f071a0188640c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-14 09:05:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4a6c95759e93287a70f0c0ae997716f878540c82","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 15:21:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4a99352a3b803f15c97447173ed6c7c68a0bc586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:23:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"119d4aa68a2d6d6598b8da93c60b690327fd0f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:08:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a224b464d7d0466aad7f6934d8050d7d : SUCCESS in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ec96058b46c4cf6be1902601a169463 : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6343d365383a4624b12aebff12f8f7e2 : SUCCESS in 4m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/dba7300237a8421da7ebf1055a9b665f : SUCCESS in 21m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c325c947d5b24d148b9aa1718793af4c : SUCCESS in 10m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e216add4ca14147b4e0fdf2c7f82159 : SUCCESS in 8m 48s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/36b3f2d967bf4b8c90a79734c5a09746 : SUCCESS in 1h 12m 54s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca2b0badd7d048ddbc562f915c62341b : SUCCESS in 1h 35m 00s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d904f27da7d34a2ba716fc57e7dca726 : SUCCESS in 1h 29m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7fc44d5659c5491eb7abd0e46fa2ac32 : SUCCESS in 1h 06m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5b4f8481b5d444dbbf914122af482c5f : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":2},{"id":"eec9d3966cfc05309c400d2a882fb64aa6ac274a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:08:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7f11025805d8da5b6f8b8066e6bd1fe0f2636e55","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-18 14:34:44.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 5fa0fca1bee859ff31357c7f0df50ee4d9b73145","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"064edacff36d949ebdd304bd404a5fdcff1f8981","revisions":{"7734d3cfe0f4c5af061e818af35c2ef18b3501be":{"kind":"REWORK","_number":1,"created":"2020-08-14 00:22:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/746248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/48/746248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/1"}}},"commit":{"parents":[{"commit":"67dcd8bd119dd7bca689f2e442dba630077515f7","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/67dcd8bd119dd7bca689f2e442dba630077515f7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 00:20:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 00:21:41.000000000","tz":540},"subject":"Remove release not file in wrong path","message":"Remove release not file in wrong path\n\nThis patch removes the release note file which was added into wrong\npath[1].\nThe removed relesae note file is not added into releasenote directory\nto avoid confusion by a note associated with wrong release.\n\n[1] 0f617ce27a99260ee03627b1fbde461a687718dc\n\nChange-Id: I8ff8886b65a323a65dd0ff9d01066c82a4116f8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7734d3cfe0f4c5af061e818af35c2ef18b3501be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7734d3cfe0f4c5af061e818af35c2ef18b3501be"}]},"branch":"refs/heads/master"},"064edacff36d949ebdd304bd404a5fdcff1f8981":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-14 02:19:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/746248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/48/746248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/48/746248/2"}}},"commit":{"parents":[{"commit":"67dcd8bd119dd7bca689f2e442dba630077515f7","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/67dcd8bd119dd7bca689f2e442dba630077515f7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 00:20:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 02:19:06.000000000","tz":0},"subject":"Remove release note file in wrong path","message":"Remove release note file in wrong path\n\nThis patch removes the release note file which was added into wrong\npath in an old commit[1].\nThe removed relesae note file is not added into releasenote directory\nto avoid confusion by a note associated with wrong release.\n\n[1] 0f617ce27a99260ee03627b1fbde461a687718dc\n\nChange-Id: I8ff8886b65a323a65dd0ff9d01066c82a4116f8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/064edacff36d949ebdd304bd404a5fdcff1f8981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/064edacff36d949ebdd304bd404a5fdcff1f8981"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
