)]}'
{"id":"openstack%2Fpuppet-oslo~749034","triplet_id":"openstack%2Fpuppet-oslo~master~I62c7d60c04b9ef0a3c74c7355c8e7a199ffc5cd5","project":"openstack/puppet-oslo","branch":"master","topic":"db-resource_type","hashtags":[],"change_id":"I62c7d60c04b9ef0a3c74c7355c8e7a199ffc5cd5","subject":"Support overriding resource type used in oslo::db class","status":"MERGED","created":"2020-08-31 11:34:48.000000000","updated":"2020-09-01 18:01:42.000000000","submitted":"2020-09-01 18:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749034-1598983302185-038e877c","meta_rev_id":"652fb1093a07d028f178a893aebd8c88448331a4","_number":749034,"virtual_id_number":749034,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 18:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-01 12:50:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-08-31 14:29:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-08-31 15:42:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-01 12:50:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 14:29:30.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-31 15:42:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-01 12:50:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-01 18:01:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c26518031c57634a593081fb7cbb960ff1406db3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 11:34:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d45e2ab37bce7cbc84c8c63f00b8a31bf7ed2115","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 11:36:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9e6f70356d04e54fa149f483d4c52ea397973220","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 11:39:45.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c563cba97dda5162b87edbe4a1fb60c9de1ad13c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:13:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5923dbf955a4091bb09d57f1b4cede3 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84c390c5a0e34f698ef78a6de51ae78e : FAILURE in 2m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9356380f61d14c1585d9b1f38e2ecdba : FAILURE in 2m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/677f2c4df3454d3dbd5245de5118b180 : FAILURE in 10m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/60affd6adccd49e1815aff5dc2d186d8 : FAILURE in 7m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e28970414812439aab4bc3ce18321b01 : FAILURE in 9m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f77db9e7161422aa8745ed3ffea6287 : FAILURE in 7m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/50927785638a47f283fa235c752a0e4b : FAILURE in 11m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06e6e9e5bfb043e78bf65e583dbc0f27 : FAILURE in 10m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c2f5252aa2a44e08efc5b793a78616f : FAILURE in 5m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cd2ea185507a4cabb5c1b8a109440289 : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/004f8a8cfcfc4af48cdcb552224907c8 : FAILURE in 13m 48s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fd8b91f2afa7483fa86ef8936db6f53b : FAILURE in 8m 38s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fa2ec9466cbe47c987130d5f560a006e : FAILURE in 8m 16s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1f2e0022da064408bf3a54a109cc3a55 : FAILURE in 8m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/dd488f65702c4e6fb5084b83f7afbec6 : SUCCESS in 6m 58s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5683c1ca8134372b4be14ee62628bc8 : SUCCESS in 7m 15s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/25fa0ee132674a1285637e512c11598c : FAILURE in 34m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3d45ddc0d6de4b5df7f8386a5cdf01463a32f657","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 12:43:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2e1ffa4b3d1f4547d62dca69e7915d814c4fa5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 14:25:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99f1abd0646b4d2faa4c8e4771a1f4cf : SUCCESS in 3m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2935fb3318b449ad8a2f2ca3cbd90739 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5aeb62ab33d24597baa2855f44e7c27e : SUCCESS in 3m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7497626d85ef4f5fa24f86efa9acd567 : SUCCESS in 8m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/00df15d25c654d54b8e9f4faaed99f9c : SUCCESS in 8m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d3f64842a364d4dbd6cc8a109b62efc : SUCCESS in 10m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c72588937a87455ca72aa2730e3762b6 : SUCCESS in 6m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/776c7022ad694bc4875296e3c751cddd : SUCCESS in 1h 02m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d3c9c7eb48a4753a3330d25ef3b00f3 : FAILURE in 53m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a97748e0a71d442585709f49d574f715 : SUCCESS in 57m 54s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/46f76462237445a4ae8f4b7f9d17e460 : SUCCESS in 36m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a6fcd94b94d14fa9bde7477870783d54 : SUCCESS in 1h 13m 26s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/521632ce06ea4a0bb6e638e6047530d2 : SUCCESS in 1h 25m 15s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ba85250424d3456bb10f3ae1579fa160 : SUCCESS in 1h 34m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/30263a4676294db1845ee6a869811e85 : SUCCESS in 1h 00m 35s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7992387f99874d0ca037fd8a9cfea3a1 : SUCCESS in 9m 49s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44fe45381d07420ba5baf5434f68ddee : SUCCESS in 7m 45s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ec70708c70354941b88af73997c64834 : SUCCESS in 1h 34m 53s","accounts_in_message":[],"_revision_number":3},{"id":"c4a0e40612f322fbe743cde8934cd95cb31711e1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 14:29:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7aa45cafc6bb72e674411703bf593293e59b9ccb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-31 15:42:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9bf6d54ba94e809cf9f805a0ba4c5956980c9f56","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-01 12:50:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5c82b509a4ccc1b62aa269c0002beb03fb96be61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 12:50:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5399e3d5da9bf057ff74af88b6326eda9e8ac6fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:01:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6e10f0ad0e348898c4f560c695e43db : SUCCESS in 3m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/595d39b273aa4efaaed243e76c0200a9 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb47fddb20ca4d5ab5d99919bf7205db : SUCCESS in 3m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0d2d2254c6004c17ad00ffe8038091bd : SUCCESS in 11m 51s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cd2e9b2f4c04468e983bc6c8f1cec950 : SUCCESS in 11m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d6394586d2b430ebd362f950c7e425e : SUCCESS in 9m 00s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/92ff923355b2409bbb27d7e0bc8ead8e : SUCCESS in 1h 22m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ec1a17ebd5844d3c85abd41465320f1f : SUCCESS in 1h 31m 09s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bf528db4fd6344da93f3a39d25bb18c7 : SUCCESS in 1h 23m 09s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/805a987fb8e04d3385ba4883d3c267f3 : SUCCESS in 1h 03m 45s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/24cc3ba094644f78bf90837808cec7e3 : SUCCESS in 11m 55s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/becd3579785d41f6b05bc8913daf3f4e : SUCCESS in 1h 34m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8c21d02bc0222d4fb8c6ea98af67a6b8ba336a82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:01:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"130282bc98af92b6d62bd1604fbbf0830813f20a","revisions":{"fbe87b08552ce75362aec9c88f4a616e4ba98907":{"kind":"REWORK","_number":1,"created":"2020-08-31 11:34:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/749034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/34/749034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/1"}}},"commit":{"parents":[{"commit":"893205bd3c4be8344b7281e5a35dc0c587cce597","subject":"Add tripleo standalone job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/893205bd3c4be8344b7281e5a35dc0c587cce597"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 11:32:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 11:32:40.000000000","tz":540},"subject":"oslo::db: Allow independent values for name and resource type","message":"oslo::db: Allow independent values for name and resource type\n\nThis patch allows specyfing resource type used to apply configurations\nin oslo::db class, so that we can use this class to set up parameters\nfor multiple databases.\n\nChange-Id: I62c7d60c04b9ef0a3c74c7355c8e7a199ffc5cd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fbe87b08552ce75362aec9c88f4a616e4ba98907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fbe87b08552ce75362aec9c88f4a616e4ba98907"}]},"branch":"refs/heads/master"},"b3764a07f5e8b63081b2965c404e816dde9985f8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-31 11:36:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/749034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/34/749034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/2"}}},"commit":{"parents":[{"commit":"893205bd3c4be8344b7281e5a35dc0c587cce597","subject":"Add tripleo standalone job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/893205bd3c4be8344b7281e5a35dc0c587cce597"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 11:32:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 11:35:44.000000000","tz":540},"subject":"Support overriding resource type used in oslo::db class","message":"Support overriding resource type used in oslo::db class\n\nThis patch allows specyfing resource type used to apply configurations\nin oslo::db class, so that we can use this class to set up parameters\nfor multiple databases.\n\nChange-Id: I62c7d60c04b9ef0a3c74c7355c8e7a199ffc5cd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b3764a07f5e8b63081b2965c404e816dde9985f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b3764a07f5e8b63081b2965c404e816dde9985f8"}]},"branch":"refs/heads/master"},"130282bc98af92b6d62bd1604fbbf0830813f20a":{"kind":"REWORK","_number":3,"created":"2020-08-31 12:43:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/749034/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/34/749034/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/34/749034/3"}}},"commit":{"parents":[{"commit":"893205bd3c4be8344b7281e5a35dc0c587cce597","subject":"Add tripleo standalone job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/893205bd3c4be8344b7281e5a35dc0c587cce597"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 11:32:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 12:42:57.000000000","tz":540},"subject":"Support overriding resource type used in oslo::db class","message":"Support overriding resource type used in oslo::db class\n\nThis patch allows specyfing resource type used to apply configurations\nin oslo::db class, so that we can use this class to set up parameters\nfor multiple databases.\n\nChange-Id: I62c7d60c04b9ef0a3c74c7355c8e7a199ffc5cd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/130282bc98af92b6d62bd1604fbbf0830813f20a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/130282bc98af92b6d62bd1604fbbf0830813f20a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
