)]}'
{"id":"openstack%2Fpuppet-oslo~756810","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fvictoria~I753851a10c92262b81e6925520bc6b36ab9863cb","project":"openstack/puppet-oslo","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I753851a10c92262b81e6925520bc6b36ab9863cb","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","status":"MERGED","created":"2020-10-08 14:45:00.000000000","updated":"2020-10-15 01:13:34.000000000","submitted":"2020-10-15 01:12:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756810-1602724342811-a0ce27bd","meta_rev_id":"28230ff384d6ba71f5f79c4ccbf9e490010b3f90","_number":756810,"virtual_id_number":756810,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-10-13 13:26:37.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 01:12:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-08 14:52:39.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-09 18:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-10-09 18:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 14:52:39.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-09 18:46:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-13 13:26:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-15 01:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f907edc0d46e3340e76a213358d6f8c6d744ac4","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-10-08 14:45:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d65f38d957c667b0543248a383ef33ef14491b5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-08 14:52:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16d0f1a206126fc941b5f5e8b1b98d7118566c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 17:55:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd4d83f954374ffa9961ecca09c5f57b : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9afd571aba346debf8d98890aa410de : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4eabf6ce205e4f60bd68e86a0b9418d3 : SUCCESS in 9m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c93cd7ca827c4cedb6acff28260e57a3 : SUCCESS in 12m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc8361c4d1ac4501977ad2cc817d739b : SUCCESS in 8m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2504c342a4ba4bf2801dfad7900f5224 : SUCCESS in 55m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a7e8e15817e4882afcc28dac433f786 : SUCCESS in 54m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/246e789fc7ac498fb85fc8e2e067acee : SUCCESS in 49m 44s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/851d5f174abb4745be9abaecbdeafbf9 : SUCCESS in 42m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4f91e62baf9346f388a9a9998ec78209 : SUCCESS in 1h 23m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2be5da5dbc1b412096e3bc29582f493b : SUCCESS in 1h 40m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/06103a5073b24834883c532340d26a7c : SUCCESS in 1h 22m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eed7c38493ea489db3caf52a8ddec6b6 : SUCCESS in 42m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d90764173fb4537aa54af790b9741ab : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b63345ea1de6c59be4f5f1579e16b72bd5a9b9c5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-09 18:46:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2f04a17faeff7361a286c7509638a1232587c1f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-13 13:26:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef09a8c31a0d652305f40368a6845ed962cefacd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:57:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9ef5883d8a5475da4e368933a874c74 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c34e5caf6bc45b88ec660b08908ee01 : SUCCESS in 3m 11s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cc6eae35b2274a76bf3b79bd3503abca : SUCCESS in 9m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e62c3cb182364e15999752126dd8c40b : SUCCESS in 6m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0637da66ab83415b8ad4c28424f71e2e : SUCCESS in 6m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ec1619e22eec40d7b97c1ac5597b92c8 : SUCCESS in 49m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b323d841c4b4d1e8a845ffc4fad9a02 : SUCCESS in 50m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/559aaefc6964415aa768085e3db9ba3e : SUCCESS in 50m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7ca382d7cbce4b96b540d827009b13bc : SUCCESS in 35m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b70db16776b84035b8b298d866a2da8b : SUCCESS in 1h 20m 32s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f6d9c02d69fc46fab07583a25c433f93 : SUCCESS in 1h 41m 42s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f729178c7dda4132835e42b8da3b5625 : SUCCESS in 1h 26m 45s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f480dbc1fc2c44ca959e396a4ed68593 : SUCCESS in 58m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50eb6bbdd70d48e697b0f9ab49023518 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2c3891074c7dd676d575dc6f07534df7d1798e0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 22:49:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50bbaf5208b3efbb08bb89f6dd65ca77645b1aa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:12:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2e9780af2e4f042cf41d0e1c38c6c9aa70a5e729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:12:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d7bfa5722ad49199685ad9f159d70ce : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/926c4fc179934c9ca098959e1a959bc5 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/54f22cdb90fe41d299dad69e04ed2909 : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61f8048e18e14a5eb3515278c45bc3b2 : SUCCESS in 8m 40s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f29d27fae3a7423ea2fa6d3f02adde13 : SUCCESS in 1h 25m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/65303e7d3ff948efa099043a6801156a : SUCCESS in 1h 43m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8fbf1f3d6b2c493a85de9558a8a2371c : SUCCESS in 1h 33m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/39f69edb90ad44dc9367d0c1faee0fbc : SUCCESS in 57m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e2b754a2f374e5d891e825d0b72a94a : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"131656f028ac58fef8e0f3724236eecdc544be7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:13:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca482f692d8c401f9a72eda65307a484 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a10cb1c7111a7fcaefb6daeaf090ed981ada97af","revisions":{"a10cb1c7111a7fcaefb6daeaf090ed981ada97af":{"kind":"REWORK","_number":1,"created":"2020-10-08 14:45:00.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/10/756810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/10/756810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/756810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/756810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/10/756810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/10/756810/1"}}},"commit":{"parents":[{"commit":"e8701f6dc1dca03bc34ca545c20892f33cdea1c0","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/e8701f6dc1dca03bc34ca545c20892f33cdea1c0"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:44:58.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:44:58.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","message":"Update TOX_CONSTRAINTS_FILE for stable/victoria\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/victoria branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I753851a10c92262b81e6925520bc6b36ab9863cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a10cb1c7111a7fcaefb6daeaf090ed981ada97af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a10cb1c7111a7fcaefb6daeaf090ed981ada97af"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
