)]}'
{"id":"openstack%2Fpuppet-oslo~759008","triplet_id":"openstack%2Fpuppet-oslo~master~I4f868e7054cc83ec09b0c97be6e341e53fd952e6","project":"openstack/puppet-oslo","branch":"master","topic":"oslo_policy-enforce_scope","hashtags":[],"change_id":"I4f868e7054cc83ec09b0c97be6e341e53fd952e6","subject":"Add support for the oslo_policy/enforce_scope parameter","status":"MERGED","created":"2020-10-21 12:03:44.000000000","updated":"2020-10-23 03:15:04.000000000","submitted":"2020-10-23 03:13:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759008-1603422834751-a6735835","meta_rev_id":"ccff983493745a0fe8a1f45d56d2aa156243ca6e","_number":759008,"virtual_id_number":759008,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-23 03:13:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-22 09:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-10-21 15:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-22 09:28:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 15:14:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-22 09:28:25.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-23 03:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fcb2a9bb8974523ca19bc9740c7567aaa3a95f8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-21 12:03:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a508cb023372d335d34525e1298dc4244e2b11f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:10:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8d25c6291ca4cac8b6f9ac71dbffa4a : SUCCESS in 5m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5cc8f60a97548eda80b3790b6390a62 : SUCCESS in 5m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5b216892cddb48aa85e5e8a4312587f4 : SUCCESS in 9m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b99150efbea4bb1aadee29c0e41765a : SUCCESS in 8m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2ab46601716442c9696ce7c9a6f012b : SUCCESS in 8m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e447150b1e69468399dd32c717b637ea : SUCCESS in 53m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a6bb707e39a47c1a809b9056b6a93ad : SUCCESS in 53m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b5f0a43e05848c48202d399e035808b : SUCCESS in 48m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5d7806ac3d3b4312a930ffbd54743cba : RETRY_LIMIT in 20m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6783ffddf48e409c85968639d5e499f7 : SUCCESS in 1h 28m 26s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/015afae5c1ba4cbdb7b101cdcda631a1 : SUCCESS in 1h 35m 11s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/035061f226f34b478e4dc8dd38ec76b0 : SUCCESS in 1h 39m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cfacc30704ee49a68c352dd9b4286153 : SUCCESS in 1h 03m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21ff5c3376214c87bd2b1b829689fa76 : SUCCESS in 5m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e13d070f6f9046b6b44afb34afaea563 : SUCCESS in 1h 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d6276c556cbad50cf0934ec233c03d9ca5ad1e24","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-21 15:14:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2b97ed9535ed1e893b3e3c900345935291b32b8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-22 09:28:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1447965180d89fa61b1f5d5b83af0172452b3bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 09:28:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"887863a1a7d725c10ddafdec4b0599899d81e435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:46:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfefe5fbbc7746c1be7a60a4cce33e15 : SUCCESS in 7m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba17ca476e194e889656f1b44cb1f275 : SUCCESS in 6m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/86b8c60703c9487b9b18bb6fbdc1936e : SUCCESS in 15m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf98bd004ec34c2294c65e6567c19c94 : SUCCESS in 9m 51s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c4fdddda447c4c9bbeacd0915fd78a3f : FAILURE in 33m 35s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8c203522ad914555b88ee738a24dc2df : TIMED_OUT in 2h 04m 56s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7f7c03f2e0bb49a8a545699c0bbcdae2 : SUCCESS in 1h 29m 42s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ef969d76afca4ad2beb24cc62417f9bd : SUCCESS in 58m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89eabcf23cd540df96657afe07bc527a : SUCCESS in 7m 04s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5d32e10b3a274fc096bcad6d4fa7620a : FAILURE in 1h 47m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9f59b02a4e11bba7892293e928a4b00508512e5f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-22 13:43:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2dbded538d2fc83883445734b8e5caf0fb81bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 15:39:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bee039d9f8914264b696a723749c80bf : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1135d153f794b56a29c49ef21f935b0 : SUCCESS in 4m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e2779abb0bc04a0587fec6292dc7e4e0 : SUCCESS in 10m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62ab45878ded4c2a83294b997af960b3 : SUCCESS in 7m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/313fcde3fc354980bafc16b099a6693a : SUCCESS in 8m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/894350874bc54065b4b326ad6e6d2f78 : SUCCESS in 1h 00m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ddc2fdae258484d8b46ef2a64a88cf9 : FAILURE in 1h 04m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8d679114cdc4fb9970c27c1b593517e : SUCCESS in 55m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/eca86d92b0534ada8535e2e6151c842c : SUCCESS in 52m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/52a5fc144543431794a087654c157615 : SUCCESS in 1h 25m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dbe9269105e746f39bd283346dcb1de6 : SUCCESS in 1h 38m 57s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc45699f136b4266859eacc75993e6ba : SUCCESS in 1h 33m 00s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f1f0ac91d7604a9b98d5d35a6ece6d95 : SUCCESS in 58m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa462422237d467faa477b6d92c67650 : SUCCESS in 5m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/365e75189779478b95fd89c07428acc7 : SUCCESS in 1h 44m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9b7bd497b165c2162cb507522b32ef72eded49ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 15:40:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a74b34eb1eeebf3525e1575096fad9bda591189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 21:11:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2eb14171a9a418bad977e5ce3c31dc0 : SUCCESS in 6m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1d16f841ecd45d493203684929b0ff3 : SUCCESS in 6m 27s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e46f74f80755435a8fc20b74d725cddf : SUCCESS in 17m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/804bcac4aa394a4490d60ddfbd4217d5 : SUCCESS in 14m 38s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a1625f5a9b1748ca8da120fe9ba35a54 : SUCCESS in 54m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d199ba3b5bc946a89183214eb6f6b573 : SUCCESS in 1h 44m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c0753ec40e1845c2abc4fc7d79fb450a : SUCCESS in 1h 40m 50s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7f5c723377684eea9e7097d0dc949c2b : FAILURE in 1h 01m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e31156d9f6274c788c0abed87db009cd : SUCCESS in 6m 45s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/51de2cffe2b64c79bd936f41f4d8b3c2 : SUCCESS in 1h 57m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fba44fb899c2715b07d0ce178736f0cbd6ed3865","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-22 21:47:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac9e2920afedabfb20ddcfeacf3fe501ff73bdf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 23:41:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55d77da3f4da4007845d9efcb7b82c06 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13f60837284f4d24882a304a947f693b : SUCCESS in 4m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c0de96bf4b3645d88bdf65816b946aaf : SUCCESS in 11m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5305cd65ee024e1d9724073392205d79 : SUCCESS in 7m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c412ffaf55644515917aea0a411e2265 : SUCCESS in 8m 18s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0335ea3c2b2f49409177b3d99e4c849c : SUCCESS in 1h 03m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8a7b93d814646c2a6da55cb22635c23 : SUCCESS in 54m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb42dff6b19c4847a8af505ffe67bf7e : SUCCESS in 51m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/525713f57c444334ad9fc479658e1508 : SUCCESS in 43m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c109cc8223fa470db2b25287884bdd4c : SUCCESS in 1h 35m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3908b1729f5640a29a166f2421466d06 : SUCCESS in 1h 30m 42s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b46cc3c1de95425aa655480824cbc54f : SUCCESS in 1h 24m 51s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/491fc0c8c8e24ea184e767676e5995bd : SUCCESS in 59m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b99333a250bc4dd4a136ae0d77071258 : SUCCESS in 5m 14s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9047ee87644b4abe88523e889be195b2 : SUCCESS in 1h 48m 04s","accounts_in_message":[],"_revision_number":1},{"id":"27950936c947a2c47fabfe0ec6925e675fc78cfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 23:41:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7076c7a6a0df4935344664c9994fcfc492bb91be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 03:13:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff194a08c2b6d6adc545792b07ff25e96397ea79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 03:13:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96c3d9263a0543568bcc05463de460c5 : SUCCESS in 7m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68b13612a28c4355b5a5a0ade7237f33 : SUCCESS in 7m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f953c46ddf5e49aa9de55782fc0b2e58 : SUCCESS in 11m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/536d6aa81d3949968d12431a99d6f892 : SUCCESS in 13m 02s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4bb727731fb94d78b20b8a5b67ce961d : SUCCESS in 58m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/58517268a55749ef8e00b9041b8cf33b : SUCCESS in 1h 51m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/25a4e8fbbf304719b332e4eb875ac8c5 : SUCCESS in 1h 29m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/db96852bbaf843f6b138c4ea1aa2f272 : SUCCESS in 56m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab179b9f5d245bbab845367ed0dd4d9 : SUCCESS in 6m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7250b070ec72422092e61ff324597486 : SUCCESS in 1h 59m 17s","accounts_in_message":[],"_revision_number":1},{"id":"406d42f4debf7b5cb48bbb2280f27daf664b99cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 03:15:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/667fe44111034439b486e1d913aab679 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21d6bb7af163af425c88ed7b8e17e46b8cc8af78","revisions":{"21d6bb7af163af425c88ed7b8e17e46b8cc8af78":{"kind":"REWORK","_number":1,"created":"2020-10-21 12:03:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/759008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/08/759008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/08/759008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/08/759008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/08/759008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/08/759008/1"}}},"commit":{"parents":[{"commit":"c737a0556f352d548bbec6ecbd18cbf3a341e08d","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c737a0556f352d548bbec6ecbd18cbf3a341e08d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-21 12:03:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-21 12:03:21.000000000","tz":540},"subject":"Add support for the oslo_policy/enforce_scope parameter","message":"Add support for the oslo_policy/enforce_scope parameter\n\nChange-Id: I4f868e7054cc83ec09b0c97be6e341e53fd952e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/21d6bb7af163af425c88ed7b8e17e46b8cc8af78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/21d6bb7af163af425c88ed7b8e17e46b8cc8af78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
