)]}'
{"id":"openstack%2Fpuppet-oslo~774120","triplet_id":"openstack%2Fpuppet-oslo~stable%2Ftrain~I415cfee6045cd904e9675b97750c5de2f1be55d3","project":"openstack/puppet-oslo","branch":"stable/train","topic":"memcached-cert-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"I415cfee6045cd904e9675b97750c5de2f1be55d3","subject":"Add TLS options to oslo.cache","status":"MERGED","created":"2021-02-04 19:23:25.000000000","updated":"2021-02-22 19:52:15.000000000","submitted":"2021-02-22 19:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774120-memcached-cert-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":774119,"cherry_pick_of_patch_set":2,"meta_rev_id":"f10c3d521f2af44b90b6d50b8a1f62c589093aed","_number":774120,"virtual_id_number":774120,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 19:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-02-21 06:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-02-22 14:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-02-22 14:52:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 20:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-15 11:02:33.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-02-15 11:02:33.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-02-19 02:48:55.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"791d1b5784a3a0c497c0325b678afd51c7fda5bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-04 19:23:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"805597885ecc35aa1530427866c61cdd83113e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 20:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45df33ec7155437e891b0bea2e660690 : SUCCESS in 5m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7817c716b5ca4771beabb21f1d461bd6 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cfc8a9c91424f6eafa7529060670973 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0754f66e05184a4182cebf43882ed94b : SUCCESS in 8m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e068a9bd6a4a4fa98de01e1b0f474755 : SUCCESS in 9m 58s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/e174d99ca59d4533a560fe91d62a7233 : SUCCESS in 9m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1abbaaee2b44dfca7e30025c758664f : SUCCESS in 7m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73ef213a23ea4168ae93d1b9c52b3cb3 : SUCCESS in 8m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/68145f5ba430421d82033136e6593185 : SUCCESS in 48m 50s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4c64aeada0144792b90cd70749f40e2f : SUCCESS in 53m 08s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/82e28a2897b84cb5abb0aa41a8beeab1 : SUCCESS in 56m 26s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/db69708240b24cf781e2258a490609ee : SUCCESS in 43m 21s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b9826ce037774de7a2dbfb5f148fcd39 : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19bd4155bd09452aab36235c622ed568 : SUCCESS in 53m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea511ec8e1ca4bcb92298cdb1df8ee0d : FAILURE in 42m 41s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5ce0e8339e694439a961677baa93cc84 : SUCCESS in 40m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/68cc403e859f48299fa8c381b5ab164d : SUCCESS in 48m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/99a7f0494b0d498ea02bdd5e7ebc9681 : SUCCESS in 1h 02m 29s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a72b697a083b431ab67f32b7a2bcad70 : SUCCESS in 56m 02s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f4068f56b7a141c39af42ecf4d186ba0 : SUCCESS in 41m 14s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8e6c56a4527c4e66809cc5bbc5d29bcf : SUCCESS in 8m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef20683c9cf240f4ad78d2d24a279219 : SUCCESS in 11m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9395d3f9c38d4d9e958d7e1bdb6853e0 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1c1f0da03533f09e19746f898e325c3dd97adacb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-15 15:10:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d316eb7759b77e2f06bec93db840e6b5a762f52","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-02-19 02:48:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ca2b6470fede35fae9d3d6c4cd69704d185fd5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 02:49:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d02bd28f9dcdf56961df7d88ca2b3205e93cd6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:54:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52eb014f80704a9ea675010359d20fc8 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7991357314941bd9f46a8dc15e708ab : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa5714bd74334ae989b261f7ab80513e : SUCCESS in 5m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fd1417f77a204b728f06c93269e41883 : FAILURE in 7m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24afcaf7216c4c1aa31e28cae5fcc63e : SUCCESS in 6m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/daa39fa3f3da4cc08f6364d21c1f8afc : SUCCESS in 56m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b345d8ad37704b86be12ef115d869b54 : SUCCESS in 57m 03s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/aff77e7e1e8e444189630cd126f481d7 : SUCCESS in 1h 01m 15s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/0ed4002af4f04b3484a07f65aa740b23 : SUCCESS in 43m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f0010571eef64f338518e440b2406ca9 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1d9e90213ca4afa906cc4eb7cc9c991 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d2eb035b6e7697f273c06ab248d058c84daf46db","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-19 16:12:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43078d5504f5cb01f1cffd42173f34e91a7f4d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:38:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/245a58ad6dc24c2fa6ebbc7a219b7570 : POST_FAILURE in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff1a22de2eaa4a8fa469d855dfd2a827 : POST_FAILURE in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65d2cabd5bd847dea2e625172263705b : POST_FAILURE in 5m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/69e6588ba81f4f47be6024ee61c9012b : POST_FAILURE in 16m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3f78ef99c8c84228bd67839828adbea6 : POST_FAILURE in 9m 44s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/bf4f7a0a23914c2b9891e1bdad4ecccb : POST_FAILURE in 9m 08s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba14c3c6484c4758972ee91565f4eed8 : POST_FAILURE in 7m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d96fff6c7a5487396a0692695a1b0a3 : POST_FAILURE in 10m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1060b8384b7545b7a26aa5670d126801 : POST_FAILURE in 55m 14s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/58e742be4232494295b47b8d13500baa : POST_FAILURE in 56m 03s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/90bb2cd1aa6f4ba89e3c017b4d53e5b2 : POST_FAILURE in 58m 50s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/5679c217f57f47e9946784612ab8f635 : POST_FAILURE in 57m 53s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/20ef704bd48d4324860ed24dffa980a4 : POST_FAILURE in 34m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd5c929c4cf54eb78b06e4c88414739a : POST_FAILURE in 29m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c694ca4c61b48459dd81f16beb7b00c : POST_FAILURE in 25m 34s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8ac3fe22256645b9b76fc4e80795996c : POST_FAILURE in 22m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/03ef7759cf4c414e9280d7d358529a01 : POST_FAILURE in 52m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/413465391a284c86880dc57ee65cf94e : POST_FAILURE in 1h 03m 45s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cab9c3a531a44d958d1dd36d085e49e7 : POST_FAILURE in 52m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/08cec024b7ff4c57b1fa205546990557 : POST_FAILURE in 40m 37s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/64ba9a41a26e40baa3463020ec4df53f : POST_FAILURE in 14m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ffda31526924b0a9c5f4ee1bf01e15a : POST_FAILURE in 8m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cac423a66234524b460275db6fd7721 : POST_FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"85a21d2df17bc4b228fce8b60486f674d3df27c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-21 06:05:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aec222a05f3bd6bfc54a7a46d0c761694a1e79ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-21 06:05:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba2e2750f0b49d716e5659dfdf6377571b32971e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 07:16:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/944145e65120438aa159e844101b9c5d : SUCCESS in 6m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e78d5721169d4fb88a095343a32576cf : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc23fe5fa994468a94afd89dae9e8fca : SUCCESS in 6m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/20be979d8c2e4b4f913663ede6fefdbe : SUCCESS in 16m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/77dbe79c84c1420d8416b52eb354dfbb : SUCCESS in 9m 27s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/40d6a1c363cb4f55b849d6c41cdc87b2 : SUCCESS in 10m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfbe167881c642c7a943b1375e8e1529 : SUCCESS in 9m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/172893e342944c40af629d9d4d6ee623 : SUCCESS in 9m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3ba51fe5d3da43f699c2924b39b473fd : SUCCESS in 53m 27s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/aeb777c55339439297ac23f257a2b478 : SUCCESS in 57m 52s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/437b5271933043949ccad702f9fe86b1 : SUCCESS in 58m 03s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/625a3838dfad423ab6b466a456918556 : SUCCESS in 44m 28s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7c62a68f936a40a58d78c2d6b2ef0204 : FAILURE in 29m 13s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6307a26075c478da0b4214a21ca29ad : FAILURE in 33m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96bd21e3705e4c29b1016d9d29e8fdfc : FAILURE in 34m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f1b0452338a44658a1f84a2bb2a51f44 : FAILURE in 32m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/06f0cd5692a8462d8ee22464cf28bf98 : SUCCESS in 50m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bed755c45ce4444ea347ea3332b09420 : SUCCESS in 59m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0537523f8426434b962aa6fb1ce1cf4f : SUCCESS in 43m 43s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2217b47fbb4043c08a67c25eb275bbac : SUCCESS in 46m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/059ca2032db54755920e25e2e84bab7a : SUCCESS in 9m 47s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/700717d41a364ba4bf3778bd2a5a37b4 : SUCCESS in 9m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaa1b0da535b43b592d60e4caf0edc51 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"3b6ba621a397045904e849f813a785492304148b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-22 14:52:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"226f43328c904df54509dfb538aa3fa4e4a13bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 14:53:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c02312ed56534c99eb25ecea72c3c4941bb00a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 19:50:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b540e681b1554ea989ca2741afec5da5 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5cf067af97f47a78249cfdc4ec437ed : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35063f327c0f419a9be093bff8424b75 : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/36ee4616e3e64b548a9deac260804257 : SUCCESS in 10m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30922780c81c49fcb8392d8866d2041b : SUCCESS in 11m 48s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1c606a4d705b45738059978273e0dd8b : SUCCESS in 1h 40m 05s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ef741e71eacc42a6a28724406070006b : SUCCESS in 53m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/92aab44cd3884696aca9f3858d20b418 : SUCCESS in 53m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/72f73b57f513443e82efc3aadda46409 : SUCCESS in 46m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4ca7e357ce754b0db5045d35e3f0cf82 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2aca2c044e44f24820a61904e96892f : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4823eeec9823b36e753e5ed95e48cf3444b26d90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 19:50:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f10c3d521f2af44b90b6d50b8a1f62c589093aed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 19:52:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f870c1203fe4f8fbb8e4213c36f3f7f : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f69fbbd4799c7236505923c631432dc3b774e88","revisions":{"7bc2904f48a6e85a6f18d2a6796785c8ace2274a":{"kind":"REWORK","_number":1,"created":"2021-02-04 19:23:25.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/20/774120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/20/774120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/1"}}},"commit":{"parents":[{"commit":"185dec2fd691393bc941eed19c4ca6bf870d048f","subject":"Consolidate python memcache bindings install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/185dec2fd691393bc941eed19c4ca6bf870d048f"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-05 13:25:09.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-02-04 19:23:25.000000000","tz":0},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/725405\n\nChange-Id: I415cfee6045cd904e9675b97750c5de2f1be55d3\n(cherry picked from commit 5a8dc2b7525d58f241cbeea1ae1631d22ce87481)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7bc2904f48a6e85a6f18d2a6796785c8ace2274a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7bc2904f48a6e85a6f18d2a6796785c8ace2274a"}]},"branch":"refs/heads/stable/train"},"7f69fbbd4799c7236505923c631432dc3b774e88":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-21 06:05:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/774120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/20/774120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/20/774120/2"}}},"commit":{"parents":[{"commit":"185dec2fd691393bc941eed19c4ca6bf870d048f","subject":"Consolidate python memcache bindings install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/185dec2fd691393bc941eed19c4ca6bf870d048f"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-05 13:25:09.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-21 06:05:12.000000000","tz":0},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nOriginal oslo.cache change:\nhttps://review.opendev.org/c/openstack/oslo.cache/+/725405\n\nChange-Id: I415cfee6045cd904e9675b97750c5de2f1be55d3\n(cherry picked from commit 5a8dc2b7525d58f241cbeea1ae1631d22ce87481)\n(cherry picked from commit a0d1e4650e8be4008628dbc68ea507e09735d2c9)\n(cherry picked from commit dc725a221af3e127e30c72694582f690fd3baf21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7f69fbbd4799c7236505923c631432dc3b774e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7f69fbbd4799c7236505923c631432dc3b774e88"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
