)]}'
{"id":"openstack%2Fpuppet-oslo~781428","triplet_id":"openstack%2Fpuppet-oslo~master~Iad9b2375cbaa66943fce4f26df863fa55d590f57","project":"openstack/puppet-oslo","branch":"master","topic":"enforce_new_defaults","hashtags":[],"change_id":"Iad9b2375cbaa66943fce4f26df863fa55d590f57","subject":"Add support for oslo_policy/enforce_new_defaults","status":"MERGED","created":"2021-03-18 11:52:49.000000000","updated":"2021-03-20 03:40:35.000000000","submitted":"2021-03-20 03:39:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781428-enforce_new_defaults","meta_rev_id":"1af80b97db5afca830a07d9722a5faea202c3536","_number":781428,"virtual_id_number":781428,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-20 03:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-19 10:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-19 13:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-19 13:47:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 14:11:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 14:46:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-19 10:35:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e936953b7be12abd2bfb4470c288606be747170d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 11:52:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"096cb34a33ff1f3b35c8b74a2ad66b6ff86d1e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 11:54:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e039348d4127d98f7a323001403305232403c6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:11:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a69c49b6faa4648bb5c00e1e3b18667 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf20a86c20064d8e9941381683d9d26b : SUCCESS in 4m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ec608d5664204a2eaa5cc18a78d01bc5 : FAILURE in 9m 00s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/625ae27327ef411bb0c9919006679d00 : FAILURE in 8m 57s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/811a3940406449e4add58ea3ae72774a : FAILURE in 7m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4be80b786434825b0da33868919593c : FAILURE in 7m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bccfd98d25c4659986c722daf98c494 : SUCCESS in 46m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba6504faf4b5452687006f950df692c0 : SUCCESS in 50m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8350c2e28e994b15be6fed417f968b92 : SUCCESS in 47m 02s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c2a233b6d2c46b1804e38dd677fefc1 : SUCCESS in 40m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1e7c8277623f436cbab14fb4a7588be6 : SUCCESS in 1h 31m 19s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/537f8a2f37264472ba4d9ae0e1635636 : TIMED_OUT in 2h 01m 49s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/06fc73ff580148008ac2efafc03fcc11 : SUCCESS in 1h 22m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7e34a5a8e61d4575b7eb497857334d7e : SUCCESS in 1h 05m 56s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e161386b942949b6a74ef343b1d2d830 : SUCCESS in 1h 11m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7908202e1284dc38f923718140ad591 : SUCCESS in 1h 41m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1bf3a1b520d8471bb05eb4167d2250f4 : SUCCESS in 1h 15m 13s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba84a22538bc42dab6590e1c3e66057f : SUCCESS in 1h 01m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f201526f8789464abea94bef5fda369d : FAILURE in 4m 50s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/02027bcbb9ea446f96764cc2a762b8b0 : SUCCESS in 1h 27m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f219710afc93d2bc024da1337d96bc268bde15b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 14:17:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dcd150c599009efb4cf7241b10325b4b8651b320","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-18 14:46:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e3effebe27fe60968f3db9609b85d791193279fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:26:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7e15f1b301545a28537d75905c11250 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2eda30ac24c47a6b8a9a56328c4289f : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e9c7a82f07a7442686f7c41db3d04b43 : FAILURE in 10m 07s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0b7a0e39382416a9069671a4841df5e : FAILURE in 10m 53s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a69628bd58524453aa7cbf6be00ca354 : FAILURE in 7m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c5d6aab96134b3ba17412fce8c84306 : FAILURE in 6m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c63e9f4959ed4a5db848e0e28e8f5668 : SUCCESS in 55m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b2e0be35f63447286c709b62260e36e : SUCCESS in 56m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2600eeae74f04d00b0c27675536d5a54 : SUCCESS in 49m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf64cb57e9c3448bb1bb5f0ba94d4242 : SUCCESS in 44m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7432b959d991412da6d3ef58855b5776 : SUCCESS in 1h 21m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/45be50c9370f44d5ab638639db87405f : SUCCESS in 1h 44m 35s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1056f13150f44b9992393493c240f5fe : SUCCESS in 1h 23m 40s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6389c0d68d724f29aeff23f180ced82d : SUCCESS in 59m 15s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5339ace327e46dababcf12d5298914d : SUCCESS in 1h 10m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c2e195530b9413db8a109f0dba33545 : SUCCESS in 1h 23m 16s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/676ceb7929e54362ae342b411bacab68 : SUCCESS in 1h 20m 32s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b5e8066127a4a9f981968591473ae02 : SUCCESS in 1h 00m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db3caaf6586546e689832b325d06f416 : FAILURE in 4m 44s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7aefeaefa3244e0890a497f9aaaac77d : SUCCESS in 1h 37m 46s","accounts_in_message":[],"_revision_number":3},{"id":"9c8ecd494a61f8d75f815da1737fc01b49e80fff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-18 23:05:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c775581bad15dcfd4389ae53ce65e552ab46006c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 00:42:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/827c74002a9f4cd59a4271b24b0fa971 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ba6bb57ad574a88b4add47f24860100 : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/de1154461c594ab3b002ab6ebb8ae8a1 : FAILURE in 9m 12s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7255983b1d64da78c1aef76ea9f8cc2 : FAILURE in 11m 24s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cb1d91e8c4749ccb62ba0af1914fca4 : FAILURE in 6m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a06a0043d95b486aa82bb976363ef774 : FAILURE in 7m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e055a1200e04d13a4db0b22387f09f1 : FAILURE in 35m 36s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e397a38118d40dbb04f2a6b9d513b5a : SUCCESS in 51m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/244aec5c396c4ece8b4ec9ab4804c84d : SUCCESS in 57m 01s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f22af0d1e2e4e688e3422a2d202e284 : SUCCESS in 37m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/27fa2d26f2a342479472f442c304bc56 : SUCCESS in 1h 21m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c3d03c76f9754ce0be3c3a557090fc3a : FAILURE in 30m 23s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d606666265c64da89fd73c7186f1ddeb : SUCCESS in 1h 28m 09s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0a350254c32744d1adc91a9ecaac8145 : SUCCESS in 1h 08m 32s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e61d561fc6d410dbf2407df2d4aa871 : SUCCESS in 1h 23m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/053054931cb74fe7bcddd0493b93d17b : SUCCESS in 1h 18m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b103d8095c8488c916c6567c267a888 : SUCCESS in 1h 17m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/df643834e50d4a6aad383f30ebc57bee : SUCCESS in 1h 05m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/764653cd0b694a6bab9bd0fc6e9690a8 : FAILURE in 3m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a105c3a6dfcb48a19307eee2f91afbf3 : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":4},{"id":"492cc15e63904b7b9e4797ffbeac4372f28af626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-19 04:17:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bf54df7d7cf2727ab24ea16d5de5ae847e80431b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 06:17:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/827ec86384284859ad5e84b13b062aec : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d22c73ac851433487efad7091f48728 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a53ecb821a4249ddbbc893365da45740 : SUCCESS in 9m 14s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a65adb5839e4c54b74858aedbe28552 : SUCCESS in 8m 34s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d12fc6d1b09f45e6a774049be8ebbffd : SUCCESS in 6m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4a42f34009348ff9380b70e2e9d556c : SUCCESS in 6m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8b79bcfb61c41fabc5fe83933ad4df8 : SUCCESS in 52m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b83b668f1144aadbd5cc6f11e0eef6c : SUCCESS in 51m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e74fb7d3a2444acb69f56b250a1cd5d : SUCCESS in 50m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a07d61753bfd45f3993121dc2a84a1fb : SUCCESS in 37m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/333d4e936fbf45eaafc6165a8396aa53 : SUCCESS in 1h 25m 13s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/70d37cf29365447b9b00ef8756955223 : FAILURE in 1h 53m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3018c8b8d91947e5b4fabb2c329ab16a : SUCCESS in 1h 39m 45s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/81946c75dce74205bd89a3fdde30fffb : SUCCESS in 1h 06m 50s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5123ba2bc9844ae5a2f6e1af8aa6eb3e : SUCCESS in 1h 16m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ebf9d9058034c879b876c5dbce71100 : SUCCESS in 1h 33m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/831f1a960a414d5d8fb3e728595d1127 : SUCCESS in 1h 21m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17267c6d5494499b8d098ebc2d8f4fb0 : FAILURE in 1h 05m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef25eec429a64d26ac5d1f3943950ebe : FAILURE in 3m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d1570fb038d04f069d58e63229c92070 : SUCCESS in 1h 40m 48s","accounts_in_message":[],"_revision_number":5},{"id":"921c785dbfe94a5525066307f27b7161c0050fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-19 07:06:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"86eb13fccdd43a857ec25586e721822d0e9262b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 08:53:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0411122db82d4b04bc0ac14da534ed1b : SUCCESS in 3m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a88b45da4998494db78f5a73d878d916 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8ffa636e148a415188cf717bf85fe92c : SUCCESS in 7m 40s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1697c7b3e4c485dbe5ee458c2b0b300 : SUCCESS in 11m 23s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f6638ebbbba4d76b5721315973d81d8 : SUCCESS in 5m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0b6914572914fccb6955bbae357321b : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e13f60d8690841be8d44650e83183957 : SUCCESS in 50m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dc02f0358ea4ac2bbe3aab0c2c8c7fd : SUCCESS in 59m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f646310b70974ba3b9dcde1559e7db90 : SUCCESS in 50m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad9211244dac49d9804b03e1e7c45761 : SUCCESS in 39m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1ea0d2b44b094297b50bfca203ae7f71 : SUCCESS in 1h 19m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b5afd78d0e3f4c8182df2f1347eb49f7 : SUCCESS in 1h 45m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1d9b702b61114f588abba751cb9a1559 : SUCCESS in 1h 27m 50s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ac32f59433f24a5dacf335f0164dc2f1 : SUCCESS in 1h 09m 25s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/df0a2c5d8be7487c976ce8a4e03251be : SUCCESS in 1h 08m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/01ab0d0abbe140e683076386935c4d50 : SUCCESS in 1h 27m 51s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e671a18bf064739bb34d01b5588f9dc : SUCCESS in 1h 17m 10s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/11dca044eaaa4f24a43201dcf001218e : SUCCESS in 59m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7358e1fcffab413a80b0249b734cbdba : SUCCESS in 4m 13s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/232f479a812f4b89a37d159397614d94 : SUCCESS in 1h 45m 49s","accounts_in_message":[],"_revision_number":6},{"id":"6d0169f3f68461639fbfed48cb22d703f6994c7a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-19 10:35:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d1a986ff3b6bc064ec14bc3ea78e1f4ca23b5de1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-19 13:47:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6b1a8ca28cbb9838759948b6c6a39b0006f25f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 13:56:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1841d934c77edb54c015717edc045a5852c643f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:32:07.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74ebe8fc29704bea9395066be01d8726 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90d7de76439848acbde117ab28a3822a : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3eabfd734d694279be96302284357963 : SUCCESS in 8m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cde9ea9e267842a7bc2176655ad9c46a : SUCCESS in 6m 04s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2f05e5058e8c4d4fa9d0e4b500241111 : TIMED_OUT in 2h 02m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fc10fc9e7dff4d8fbd7756273b5022c6 : SUCCESS in 1h 47m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/98c09cd733ee43c38f8e0e2812da9d9a : SUCCESS in 1h 28m 38s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bed469ef2d2e4cbaa8f071291e17e056 : SUCCESS in 1h 16m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8671e07c021a4f7888e6beee90298598 : SUCCESS in 4m 24s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/27d044ce091b4c0398b02f003e411971 : SUCCESS in 2h 11m 09s","accounts_in_message":[],"_revision_number":6},{"id":"1beec615261034991368764c865a1296164ca430","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-19 23:28:35.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"35a784bee3ac27cb6cfa36f308ba56cb0fdcd073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 01:20:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9d20288da2040f0bdaf17b64f0aa4c5 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7c2f5135bbe40ae8e70edb6fda76ce6 : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/b24de6e4599947759d0cc04e6e4a77d8 : SUCCESS in 15m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60e4463fb7ec46ca906f37cb26317285 : SUCCESS in 5m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f6866d6e809420da37ea9dbc1ef31e2 : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f727eff6f46f40c9a814f220cedd3888 : SUCCESS in 56m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c5a6c93c2df473a8d8344a3921fc45e : SUCCESS in 53m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf204ffbea844163acf9eb065835d414 : SUCCESS in 48m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f173154fe3af4b4dab920432932fa423 : SUCCESS in 47m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17e891ca872c4d9bb003ce947fd8e28f : SUCCESS in 1h 16m 55s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ece9b5239629463f8fac54e133c15c73 : SUCCESS in 1h 45m 23s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4be42dc0c534a35b989cc3f3085662e : SUCCESS in 1h 26m 30s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c36bc69398884a52b2d3c872d0e3461f : SUCCESS in 1h 09m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28dc1e73030645918a707d7e2523dd60 : SUCCESS in 4m 19s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1140e786c5dc4c8b9f6d0db2bc8cf0bf : SUCCESS in 1h 46m 10s","accounts_in_message":[],"_revision_number":6},{"id":"d2e04a3b5cea16a09788678da2c4e9e831755781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 01:21:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"510d504edd02de58a5530d3461ebbb97a5993f00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 03:39:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9ace6fa7fe341238e7bb3f47dfa6c3e : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ca5be1fb1254d80b7a00f68a2620d55 : SUCCESS in 5m 03s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae6eac98339147ca9ead798cf2f8abe9 : SUCCESS in 10m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34c2842f769a4cc691fab3345b5b88cb : SUCCESS in 9m 20s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcd0b16c136a4b9f82bf088ff320cd8b : SUCCESS in 1h 00m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b514c2be7c140109cac2aaa7754d180 : SUCCESS in 1h 12m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f3f6ec63e82482baaf015ebba90e35d : SUCCESS in 1h 18m 33s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/84809bee27804c049dec2e5733ab7c4f : SUCCESS in 1h 05m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaa872c255824836b07d790bc7dcb98e : SUCCESS in 3m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/56a2f149f0be4d7ba59a990b4ef06118 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":6},{"id":"4c672d623b6df7fdf6f111fa92bc0e779c6ee997","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 03:39:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"1af80b97db5afca830a07d9722a5faea202c3536","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 03:40:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01937a4395f1404d95074fc81466adc1 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"67608ac838d98156fd2eca0a7a3a28710837b108","revisions":{"3ed74c43620601b6656fe24c564c8a88df262c08":{"kind":"REWORK","_number":1,"created":"2021-03-18 11:52:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/781428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/781428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/1"}}},"commit":{"parents":[{"commit":"0faa00bd96642e8be698f5f8a5e7cd372a6d65d6","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0faa00bd96642e8be698f5f8a5e7cd372a6d65d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:52:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:52:35.000000000","tz":540},"subject":"Add support for [oslo_policy] enforce_new_defaults","message":"Add support for [oslo_policy] enforce_new_defaults\n\nChange-Id: Iad9b2375cbaa66943fce4f26df863fa55d590f57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3ed74c43620601b6656fe24c564c8a88df262c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3ed74c43620601b6656fe24c564c8a88df262c08"}]},"branch":"refs/heads/master"},"028c92e2dee9339a7b88db99993a576730b57e4c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-18 11:54:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/781428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/781428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/2"}}},"commit":{"parents":[{"commit":"0faa00bd96642e8be698f5f8a5e7cd372a6d65d6","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0faa00bd96642e8be698f5f8a5e7cd372a6d65d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:52:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:54:45.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nChange-Id: Iad9b2375cbaa66943fce4f26df863fa55d590f57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/028c92e2dee9339a7b88db99993a576730b57e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/028c92e2dee9339a7b88db99993a576730b57e4c"}]},"branch":"refs/heads/master"},"ace3dd9f5a3dd30a6ce39f865f9209d8e20726a3":{"kind":"REWORK","_number":3,"created":"2021-03-18 14:17:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/781428/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/781428/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/3"}}},"commit":{"parents":[{"commit":"0faa00bd96642e8be698f5f8a5e7cd372a6d65d6","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0faa00bd96642e8be698f5f8a5e7cd372a6d65d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:52:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 14:17:07.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nChange-Id: Iad9b2375cbaa66943fce4f26df863fa55d590f57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ace3dd9f5a3dd30a6ce39f865f9209d8e20726a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/ace3dd9f5a3dd30a6ce39f865f9209d8e20726a3"}]},"branch":"refs/heads/master"},"daa8e85f5e09fc7bc808807ffa0860f6fd624b62":{"kind":"REWORK","_number":4,"created":"2021-03-18 23:05:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/781428/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/781428/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/4"}}},"commit":{"parents":[{"commit":"0faa00bd96642e8be698f5f8a5e7cd372a6d65d6","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0faa00bd96642e8be698f5f8a5e7cd372a6d65d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:52:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 23:05:08.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nChange-Id: Iad9b2375cbaa66943fce4f26df863fa55d590f57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/daa8e85f5e09fc7bc808807ffa0860f6fd624b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/daa8e85f5e09fc7bc808807ffa0860f6fd624b62"}]},"branch":"refs/heads/master"},"73d6ae80c98fcb66ea92ed22197e001bab1d457b":{"kind":"REWORK","_number":5,"created":"2021-03-19 04:17:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/781428/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/781428/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/5"}}},"commit":{"parents":[{"commit":"0faa00bd96642e8be698f5f8a5e7cd372a6d65d6","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0faa00bd96642e8be698f5f8a5e7cd372a6d65d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:52:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-19 04:17:48.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nChange-Id: Iad9b2375cbaa66943fce4f26df863fa55d590f57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/73d6ae80c98fcb66ea92ed22197e001bab1d457b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/73d6ae80c98fcb66ea92ed22197e001bab1d457b"}]},"branch":"refs/heads/master"},"67608ac838d98156fd2eca0a7a3a28710837b108":{"kind":"REWORK","_number":6,"created":"2021-03-19 07:06:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/781428/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/781428/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/781428/6"}}},"commit":{"parents":[{"commit":"0faa00bd96642e8be698f5f8a5e7cd372a6d65d6","subject":"Prepare Wallaby M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0faa00bd96642e8be698f5f8a5e7cd372a6d65d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-18 11:52:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-19 07:06:00.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nChange-Id: Iad9b2375cbaa66943fce4f26df863fa55d590f57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/67608ac838d98156fd2eca0a7a3a28710837b108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/67608ac838d98156fd2eca0a7a3a28710837b108"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
