)]}'
{"id":"openstack%2Fpuppet-oslo~784059","triplet_id":"openstack%2Fpuppet-oslo~master~Id183859e7cb936e1f70abb1bb26e071dd19b82a5","project":"openstack/puppet-oslo","branch":"master","topic":"bug/1922044","hashtags":[],"change_id":"Id183859e7cb936e1f70abb1bb26e071dd19b82a5","subject":"Do not convert policy_dirs to comma-separated list","status":"MERGED","created":"2021-03-31 10:03:24.000000000","updated":"2021-04-06 00:43:32.000000000","submitted":"2021-04-06 00:43:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784059-bug/1922044","meta_rev_id":"1ea3daf500a35f22ecbbb1bf7ca66079b5d88e35","_number":784059,"virtual_id_number":784059,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 00:43:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-31 12:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-04-05 21:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-04-05 21:29:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 11:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 12:59:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-04-05 21:29:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"030ac6beb22e8a516ff3a4bad61bd145d4836aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-31 10:03:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c194b89241ccc295e27f8b7def3f1acca67797f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-31 10:04:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8f7e6b7f2b104d98c9a4db01b162e45c8df2819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-31 10:07:52.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"24c6c45aa064d595fea0a4f6d8e933d3879c87c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 11:42:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22d6b24c72f44fada13afa7c3db01c83 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9a603eb17d84db0b2280782d62f9f4f : SUCCESS in 4m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e110fe7dfb9946c7b56d4b8402e8afae : SUCCESS in 9m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f1708e531404bd098dc099511492035 : SUCCESS in 8m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb8b65588f554a1782ada3eb84e0d2a2 : SUCCESS in 6m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87eb6165c3684d9d9a0bef5825df1f6c : FAILURE in 39m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/574bbcdbf0e94a5db66e5fae88ae4e1b : SUCCESS in 50m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e1bf976ed8f4f0096c05d0aedff3642 : FAILURE in 27m 54s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/583f7eefd62f4bffaaebfb0d8683c6fb : SUCCESS in 43m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2b3d4d61e4847bc99a0fee8100aeb74 : SUCCESS in 1h 15m 15s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4899441ca424934849310e13ba13123 : SUCCESS in 1h 27m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24fe4d5d1c20479f92186f2eeb37cff5 : SUCCESS in 1h 17m 57s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4196f16c39a461480ac728ba6fa95cb : SUCCESS in 56m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f5b3967a10674f558c297129e163851d : FAILURE in 21m 08s","accounts_in_message":[],"_revision_number":3},{"id":"dfca7ba3b57f08b8dacbebf0f71551a747a15ac3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-31 11:44:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f435ccf4b30bef35969cd1e1793c7158f53661db","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-31 12:59:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f362f2109ab83edc56bc26de57e9c7a6262869b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 13:24:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ecc1993aecd45389aa436d0557f4f1c : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/947c9171687841b68b320bbf2c22a379 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/59d8838ab0e6429d94f442e0ce7ab0c6 : SUCCESS in 11m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36d47585ab7142289efad581e6f9ef9b : SUCCESS in 6m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faaf2efdecb44e629adb579b5f573172 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a669201d3de949c4823339838880d04f : SUCCESS in 55m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/664eb572ea4440dc85862b37f105a3aa : SUCCESS in 1h 25m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/170816d3e56143daaa34407fb4de4f36 : FAILURE in 30m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd5441c00495451c921f49153fe7356e : SUCCESS in 39m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/071dcf74a2134ef1a475d070c7175196 : FAILURE in 27m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91160361ebd943a78c96f7bfdc7e32e2 : SUCCESS in 1h 25m 55s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ac2323516a94efaaf802bdd39903b2d : FAILURE in 27m 35s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bccd19100735440c8e2a91892e65b76a : FAILURE in 15m 08s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/35cab750cf7e46bf9e87239e1fa7dde8 : SUCCESS in 1h 35m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ef59114b1f9559ce51da1ae45e7dd988fe95267c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-31 13:29:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"df218ef5e93b4d80f9e726641a28448effbe6846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 15:38:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e822f032cc8e4652970d762cf2cc6635 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c19edfbe6df24ef3b29a82f1bf38e741 : SUCCESS in 4m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b436859c82ad471d8ddecf4b6cf29ca5 : SUCCESS in 10m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a1949bcd3044e62a1b3f0c3c6e86ea2 : SUCCESS in 6m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/545ac282fdc045babbc420de501266a3 : SUCCESS in 7m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c433ef9b0fe4a98b30c9936fb029fe3 : FAILURE in 41m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/218155867e3945c7afb58e68206f6af6 : SUCCESS in 45m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0706610dd4c54a6d8e6b86558d0b751f : FAILURE in 34m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00d0882add96425fbe5f773ab5e09392 : SUCCESS in 37m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e708c436f49451eb65a90fbf5195eb6 : SUCCESS in 1h 36m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91cb44490d644ad3a79f1e7708e76d0f : SUCCESS in 1h 31m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a188b147ee7044999bcd401495cc5343 : SUCCESS in 1h 17m 18s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c8028f3eb67441fa2735a36f9817abb : SUCCESS in 1h 09m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7fdf6e7bb606403291fa5bec2c3d4dcc : SUCCESS in 1h 56m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a78de7ffbcff3f1f09093267124c3d3051c470f2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-05 21:29:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b347d7034b89769edc22185ec5a54b960aab7515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 21:30:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"073b2bddd83cb79fe62c1dca2ee8f2673ca3062b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 00:43:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e93aac108bc4251aab52b6fc2567832 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0008874fc5bc487d831dd4ca1e4b5834 : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/86a54a35ea1344ae9e7ed37987e5857a : SUCCESS in 8m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12016f340f7e4e5ab27623fe90f3cd1c : SUCCESS in 6m 56s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/427c09ef01444d56b3238b223d77f811 : SUCCESS in 1h 24m 50s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/77f48c78692146c5914e86ab5e693e0a : SUCCESS in 1h 41m 17s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a63559691924d648bc6e3f040cdf0d3 : SUCCESS in 1h 12m 51s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/12fdc8f0d0d648cfb770334829802980 : SUCCESS in 1h 06m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/38138e4da01a4bb68e6eeef8723cae53 : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":3},{"id":"1ea3daf500a35f22ecbbb1bf7ca66079b5d88e35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 00:43:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3ae2625ea803b9d8d2522819323a61abd4d67435","revisions":{"c294951f3a2df1b276ad47192c3d3127151f32d2":{"kind":"REWORK","_number":1,"created":"2021-03-31 10:03:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/784059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/59/784059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/1"}}},"commit":{"parents":[{"commit":"5873f2e89415cc700051aa6af51eec9f358bff1b","subject":"Merge \"metadata.json: debian 11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5873f2e89415cc700051aa6af51eec9f358bff1b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-31 10:02:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-31 10:02:19.000000000","tz":540},"subject":"Do not convert policy_dirs to comma-separated list","message":"Do not convert policy_dirs to comma-separated list\n\nThe oslo_policy/policy_dirs parameter is not ListOpt but MultiStrOpt,\nthus it expects multi line configruation instead of a comma separeted\nlist.\n\nChange-Id: Id183859e7cb936e1f70abb1bb26e071dd19b82a5\nCloses-Bug: #1922044\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c294951f3a2df1b276ad47192c3d3127151f32d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c294951f3a2df1b276ad47192c3d3127151f32d2"}]},"branch":"refs/heads/master"},"4de2409ad7a4a5c32f4c686cb599353e5c79cedb":{"kind":"REWORK","_number":2,"created":"2021-03-31 10:04:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/784059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/59/784059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/2"}}},"commit":{"parents":[{"commit":"5873f2e89415cc700051aa6af51eec9f358bff1b","subject":"Merge \"metadata.json: debian 11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5873f2e89415cc700051aa6af51eec9f358bff1b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-31 10:02:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-31 10:04:11.000000000","tz":540},"subject":"Do not convert policy_dirs to comma-separated list","message":"Do not convert policy_dirs to comma-separated list\n\nThe oslo_policy/policy_dirs parameter is not ListOpt but MultiStrOpt,\nthus it expects multi line configruation instead of a comma separeted\nlist.\n\nChange-Id: Id183859e7cb936e1f70abb1bb26e071dd19b82a5\nCloses-Bug: #1922044\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4de2409ad7a4a5c32f4c686cb599353e5c79cedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4de2409ad7a4a5c32f4c686cb599353e5c79cedb"}]},"branch":"refs/heads/master"},"3ae2625ea803b9d8d2522819323a61abd4d67435":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-31 10:07:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/784059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/59/784059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/59/784059/3"}}},"commit":{"parents":[{"commit":"5873f2e89415cc700051aa6af51eec9f358bff1b","subject":"Merge \"metadata.json: debian 11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5873f2e89415cc700051aa6af51eec9f358bff1b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-31 10:02:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-31 10:07:37.000000000","tz":540},"subject":"Do not convert policy_dirs to comma-separated list","message":"Do not convert policy_dirs to comma-separated list\n\nThe oslo_policy/policy_dirs parameter is not ListOpt but MultiStrOpt,\nthus it expects multi line configruation instead of a comma separeted\nlist.\n\nCloses-Bug: #1922044\nChange-Id: Id183859e7cb936e1f70abb1bb26e071dd19b82a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3ae2625ea803b9d8d2522819323a61abd4d67435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3ae2625ea803b9d8d2522819323a61abd4d67435"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
