)]}'
{"id":"openstack%2Fpuppet-oslo~790596","triplet_id":"openstack%2Fpuppet-oslo~master~I62dd8489c7a5c366fcf729cf32fcb18012e7b727","project":"openstack/puppet-oslo","branch":"master","topic":"puppet-lint-latest","hashtags":[],"change_id":"I62dd8489c7a5c366fcf729cf32fcb18012e7b727","subject":"Fix lint error with the latest lint packages","status":"MERGED","created":"2021-05-11 09:07:04.000000000","updated":"2021-05-12 03:47:52.000000000","submitted":"2021-05-12 03:47:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790596-puppet-lint-latest","meta_rev_id":"356908e44cb1d4161a317e5ea798208c33499dff","_number":790596,"virtual_id_number":790596,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 03:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-05-11 09:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-05-11 18:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-05-11 18:51:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 09:44:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-11 10:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 18:51:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e7e3bcfc58aced21e7e0bfbdc0bbb19a746a5e99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 09:07:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9a5b2cebe7037fb1fc6c00f12dcc8ca6f7325e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 09:10:30.000000000","message":"Topic lint removed","accounts_in_message":[],"_revision_number":1},{"id":"c8b1d8779742145c27f9bfcbacf8322a38bdaebe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 09:10:33.000000000","message":"Topic set to puppet-lint-latest","accounts_in_message":[],"_revision_number":1},{"id":"623f6c3d25a5ea8d9649f958df13acbf909f204e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-11 09:44:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb9f61f0f33728cbced0ba5f0dfcb522449e4f9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 10:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/031a1b667c574208a7e9a6a7de175ebf : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1b28b498a7946ac9189b9e71f905312 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/719e605bc2544d3bba7405c672dbf8b3 : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/08e209b5cbc6496eaee9cb2d2fdc09aa : SUCCESS in 8m 51s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b60c356fb8774cb9bb246cfdbaaf1e7f : SUCCESS in 8m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4e2f9607c0b49e2b90b7f00ce08eb17 : SUCCESS in 7m 03s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b0122ed090a4ebfbddb60c0288e924e : SUCCESS in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f1f6cc586fb428caac4a6629f7bcc65 : SUCCESS in 7m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7204bc0918804f05a5c84daf63a1a428 : SUCCESS in 56m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8164c93b8d9442128b38aec08ea9fd28 : SUCCESS in 57m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f696ce8ecbd4ae0b6b6ace2138dd693 : FAILURE in 37m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38d868166ae14786a0ec2b05fddaf35b : SUCCESS in 36m 47s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9770d113b5d14e8581b2e69c511a655a : SUCCESS in 22m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb22fd67fa1046cfba90d73bf436ba4a : SUCCESS in 1h 16m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/061b08bf553444cab552d25e01efbbf8 : SUCCESS in 1h 27m 13s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/150848312f754c5c83985fcd1aeecf2e : SUCCESS in 1h 29m 37s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/09662d217e9b49448989af7bca9d114d : SUCCESS in 1h 09m 19s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1aba248245543f6be069682e5313421 : SUCCESS in 22m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1adeb4bedaf48c9958189a5fdabd579 : SUCCESS in 1h 12m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec87ff7ac8644ee1a5c36356dd7d91e9 : SUCCESS in 1h 37m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ff063e6c4a2432786512291874db63c : SUCCESS in 1h 15m 09s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7709d185e6d4e90aa2dc75e90770c8b : SUCCESS in 1h 07m 32s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0f5418c7bdd74ac8b3453b1af444d4a3 : SUCCESS in 1h 38m 03s","accounts_in_message":[],"_revision_number":1},{"id":"739cadb94b3e19b81038a3f6467574e724ee49a5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-11 18:51:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"152ae64d34c2563f4d55275aad02d836ccdc6a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:52:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31218120b93a11a6b8146f10f8a345b1a6aca9f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 21:10:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6353a41263c9400f8f02ce04f4c23f59 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b83e2219031425c8c66f1a35acf82bb : SUCCESS in 5m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bddf4298b7f43f09c1a354510ddfd8b : SUCCESS in 3m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c78aa80efce84a96b0ebdbc8e8aaed8b : SUCCESS in 13m 28s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5b5030a74504ab5a0b0fca79e97002e : SUCCESS in 14m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/982e28938aab4de1a1f8c40929cf83fa : SUCCESS in 9m 57s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c1b70a791ea478aa3f9ee2c653ebaf1 : SUCCESS in 7m 42s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5318f3976bbb4868a5d21c48bb3e9b48 : SUCCESS in 1h 49m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/58800b30f2cc4a0b99fe9f119b2ae0ec : SUCCESS in 1h 14m 39s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/adf7231a7bf24037b7be4ff3cc444217 : SUCCESS in 1h 23m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/86d84bd46e054e6cb8c2480c0052d479 : SUCCESS in 55m 01s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a67514237dda40e4b0702ccb1c91d66c : FAILURE in 2h 08m 38s","accounts_in_message":[],"_revision_number":1},{"id":"55510e242efa7c777e570bd89284f5ad07235736","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 23:00:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ec5eb904093525b525e8195408b086ea8336f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 00:58:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2b5d6450bb948bea7ea7ce523b5c267 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0610b5c6f396471bbadfb20bdeec79d9 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d51d0b3ccb84756b491eeb42eabf324 : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/703fe7e81198482ab01795b190f618e0 : SUCCESS in 10m 24s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/baeba120142d4d0cad34bc2d880dcbfa : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0394bf6be9d4ecaa154b18d98ee08c3 : SUCCESS in 7m 44s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9eab51a4c78c4e88bd6c7e02406034b3 : SUCCESS in 7m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b22bc7f426c40a2a7717c0225a5ac44 : SUCCESS in 6m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b6d28554ff74f01b173a7bab96cd404 : SUCCESS in 54m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9bd5e4365af43c0b49fb7999168af0e : SUCCESS in 56m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e414816d00264f9f8cfbd3345b5c4c3a : FAILURE in 26m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50f48eb3515f4f55a45736b71ed71a7c : SUCCESS in 40m 41s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b271c54e9df24c6c8fa25d06d4a6343f : SUCCESS in 22m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4980dab2c9264591a775debf8ede0735 : SUCCESS in 1h 26m 59s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/221e959031794ab9bd7a28d334bb8ef3 : SUCCESS in 1h 36m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/12adfee6af664b63b6a7f70ade0773de : SUCCESS in 1h 18m 42s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb00d1348105425683df7a340a731674 : SUCCESS in 58m 16s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0aced7c172c4469c97d0292cf0ad6f72 : SUCCESS in 24m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/770787c1dde64766b948ef4530fe5db4 : SUCCESS in 1h 26m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd0de6f154e44d1eba13e48022f354ec : SUCCESS in 1h 38m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d5ab3dd741842e7bd746b49cdeb3de4 : SUCCESS in 1h 23m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31c239b8275c47b2a2a9493ef241fba0 : SUCCESS in 1h 00m 44s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2bf9c82efee840fe91e97706bfbdb03c : SUCCESS in 1h 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"075818aca8ea7eee744e8d22e215a0e2120c8391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 00:59:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4decc850b74e28ede910011ba48c045b09bf85ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:47:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6feb369f5a5440f6b71de2fecfd3831d : SUCCESS in 5m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f839fbcd5c8407e9a569e8ab3d3b91f : SUCCESS in 4m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38f5aeed5dc945a6af6c3c63efea7037 : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/04242a8c799c479f84ab6306af6cba6d : SUCCESS in 12m 51s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/77569716517a4542bbf62fb57fcd9c83 : SUCCESS in 10m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b26263d41d94db281db4dd0c62d62b3 : SUCCESS in 7m 13s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ae0fd6835d14a91be449d3d868cd7b3 : SUCCESS in 9m 03s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff6daee542b64ee5a8e928b97288be0e : SUCCESS in 1h 02m 35s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/20a2da1fba5a43ed9aaacc4f188f2f93 : SUCCESS in 1h 40m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f365926da1143adb18aaacdcda2ae7e : SUCCESS in 1h 04m 01s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7815dac1b8e433b8709ce5eecfef405 : SUCCESS in 1h 03m 06s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/abe13d17d4514a59a61beed235c8bd39 : SUCCESS in 1h 44m 03s","accounts_in_message":[],"_revision_number":1},{"id":"356908e44cb1d4161a317e5ea798208c33499dff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:47:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9fe90e35c4618063a16515aceb6a91cf4e35035","revisions":{"d9fe90e35c4618063a16515aceb6a91cf4e35035":{"kind":"REWORK","_number":1,"created":"2021-05-11 09:07:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/790596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/96/790596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/96/790596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/96/790596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/96/790596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/96/790596/1"}}},"commit":{"parents":[{"commit":"c9e80026390fbf3b448f5a22ef0d4c5195497f45","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c9e80026390fbf3b448f5a22ef0d4c5195497f45"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 09:06:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 09:06:30.000000000","tz":540},"subject":"Fix lint error with the latest lint packages","message":"Fix lint error with the latest lint packages\n\nThis change fixes the following lint error which is detected since we\nremoved pin of lint packages.\n\nmanifests/init.pp:7:WARNING: No matching class parameter for\ndocumentation of oslo::sample_parameter\n\nChange-Id: I62dd8489c7a5c366fcf729cf32fcb18012e7b727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d9fe90e35c4618063a16515aceb6a91cf4e35035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d9fe90e35c4618063a16515aceb6a91cf4e35035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
