)]}'
{"id":"openstack%2Fpuppet-oslo~790861","triplet_id":"openstack%2Fpuppet-oslo~master~I4478d209c23e7468e6f36fe417acc01d7b986938","project":"openstack/puppet-oslo","branch":"master","topic":"listopts","hashtags":[],"change_id":"I4478d209c23e7468e6f36fe417acc01d7b986938","subject":"Cors: Accept array for ListOpts","status":"MERGED","created":"2021-05-11 23:23:18.000000000","updated":"2021-06-28 11:10:07.000000000","submitted":"2021-06-28 11:10:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790861-listopts","meta_rev_id":"76bb5aaa0762795cde16b5e90cbb9dcc6eb46a49","_number":790861,"virtual_id_number":790861,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 11:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-17 15:52:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-06-28 09:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-06-28 09:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 01:11:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 15:52:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-28 09:02:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8e9e07d2d0ea5a63e33b8c9273a426a7a1c19039","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 23:23:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44518193900ef56860711f158f6cee0caa585887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 01:11:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ee04e7d8e8c43cd8e6b0959ce677282 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68bf36e265bb4e67afd250a8d931113d : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5377c180c24b461e91e1d8700a731ec4 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e64955d6fa684f09ba3245642bb73d70 : SUCCESS in 10m 19s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/26555483928f4f3383294f45c325a935 : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a32cec1f57845beb7d0dae0789d6a1b : SUCCESS in 6m 28s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ccc37e617304fe8a985e210f83efeba : SUCCESS in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49d3389b015c40f184c5ba341a5c66cf : SUCCESS in 7m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97edd4e85ee04d53bcd32e1e3c2c37bd : SUCCESS in 54m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9412f4cb0e3d4c76ae973e2743b94c10 : SUCCESS in 52m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f90f1aae6ec4e528ba5b68b12673426 : FAILURE in 27m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1d1de7b5ff54ff4ba28f02ada6db6ab : SUCCESS in 43m 48s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bd8fbdefb9d45908cd5cd44e91b7332 : SUCCESS in 18m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a89415482164968a7d5443dda8943db : SUCCESS in 1h 12m 53s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9d03693f2e94b5995fa988d9e245006 : FAILURE in 58m 11s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4b8493c7f724c94b566bd7db6e4e824 : SUCCESS in 1h 01m 57s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e17d36b83f764acbab9e4238b0a86c10 : SUCCESS in 1h 00m 11s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ecf655fb4b9474fb8e4f615489f37a0 : SUCCESS in 20m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d27086702074506852dccbcf0966986 : SUCCESS in 1h 21m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/94512a00a6054857812685b754082de6 : SUCCESS in 1h 25m 58s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/64f782d7e778451ca2d05a11daaadc17 : SUCCESS in 1h 26m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d9ec9b672034b1ab578a32432ed542a : SUCCESS in 50m 59s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7529deb292d94d6b98b5d3b1bc7c2bbb : SUCCESS in 1h 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a6eb971f4f5c0370ab6adce77373a46fb6e44221","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-12 01:59:35.000000000","message":"Topic lint removed","accounts_in_message":[],"_revision_number":1},{"id":"d5bf13752907bcfbd0459acf098b0f76079aec3b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-12 01:59:38.000000000","message":"Topic set to listopts","accounts_in_message":[],"_revision_number":1},{"id":"98f977bc2eb802ab83f296c04d0499cdac116c9d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-12 01:59:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c1c45093390ee3f978c6dee1deb874de0929136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 03:56:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58fafe453e5c427db99f41c5294b146a : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5569952fbeaa49aeb17c02de730a0f51 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/639f060d377c410899c65ee8995334ec : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/654efcea280e4ed8a9f8bcfcbdc0d967 : SUCCESS in 10m 02s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/632ce68e510b4e56a9511def317e19b1 : SUCCESS in 11m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fe061aac022466982b10d9cab3d923f : SUCCESS in 10m 41s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b978db20f4294b5c84e740bf9bd7f33e : SUCCESS in 7m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc45935c451c46eaa9d1995e7e968ebe : SUCCESS in 12m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c700386a24dd42ca82d0d127fad1f85e : FAILURE in 41m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2df0f902de804bbb9dc36d9a4f4d2ef4 : SUCCESS in 54m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef2cd29409ee4220b5d9d7f453f75d87 : FAILURE in 32m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38748165547c4d39bbbb74d516c9851f : SUCCESS in 58m 55s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f03582f9a8614bab8ae8ca31f3282a71 : SUCCESS in 26m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fea7e98cc8f4d26b087c0c249242aad : SUCCESS in 1h 20m 58s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb6ff95e41cb42c7b3c688c714f5195d : SUCCESS in 1h 31m 45s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/581db0b4ffc341429ba627161cf4d9ca : SUCCESS in 1h 24m 13s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e3cf68978514dd880fe2c87581f157e : SUCCESS in 1h 04m 11s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fb1ca3f52c04b63b84d930585e058b9 : SUCCESS in 23m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5286f47422d43c1a913acd198ea1ba5 : SUCCESS in 1h 15m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c20cd944431540a884acfba648e5c4d2 : SUCCESS in 1h 50m 17s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f1a406433284745970bd3f0daa09b3c : SUCCESS in 1h 30m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/08e62da5a8874ea3926e0e80a345480e : SUCCESS in 1h 06m 27s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/872a8604f5a64a0b802e2a257acc8969 : SUCCESS in 1h 44m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a892f36762d07f79836735b598b26224b01d52d1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-17 15:52:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d9c4dbab112e3dc9c1b41af8f75e86010a952004","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-28 09:02:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c2f43f37f0a8253f1a23d2d4c70cef1d61919b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 09:02:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ffe008e9300b268694a555ef5a6cf759e8c18d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:10:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/381800e3e6c74c0792cb7dc0aa4d9bab : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe1eb224c0e94de9813dc8bf4066721b : SUCCESS in 6m 12s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9079bd223e594020841b478cd67552bc : SUCCESS in 6m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7159a7853c434c55b48bdc228231ce07 : SUCCESS in 14m 26s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1579aee62798464d9b33f209a01096a5 : SUCCESS in 13m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b647888912c4c1f9413ef0e9b7c3bae : SUCCESS in 10m 07s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3979c2cc0dc4b7fb7bed764eef07a64 : SUCCESS in 10m 08s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a8be6cad271415ba7349364258ae3ca : SUCCESS in 1h 13m 27s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3648885a960446c9d42b6f906c19256 : SUCCESS in 1h 29m 45s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99630737302b4bfc9dd70c93431633c6 : SUCCESS in 1h 20m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f90206b04d74474fb6333cc68d93cbb7 : SUCCESS in 1h 04m 23s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9cc8f5c11a28408598a3d10d58cb87fc : SUCCESS in 2h 01m 22s","accounts_in_message":[],"_revision_number":1},{"id":"76bb5aaa0762795cde16b5e90cbb9dcc6eb46a49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:10:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"769ff6dec9cbc9fe1c48b6493027407b37217f96","revisions":{"769ff6dec9cbc9fe1c48b6493027407b37217f96":{"kind":"REWORK","_number":1,"created":"2021-05-11 23:23:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/790861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/61/790861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/61/790861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/61/790861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/61/790861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/61/790861/1"}}},"commit":{"parents":[{"commit":"d9fe90e35c4618063a16515aceb6a91cf4e35035","subject":"Fix lint error with the latest lint packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d9fe90e35c4618063a16515aceb6a91cf4e35035"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 15:00:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 15:01:47.000000000","tz":540},"subject":"Cors: Accept array for ListOpts","message":"Cors: Accept array for ListOpts\n\nThe following four parameters of cors middleware are ListOpt, thus\naccepting array for these options allows users to define parameters\nin a similar format.\n\n - allowed_origin\n - expose_headers\n - allow_methods\n - allow_headers\n\nChange-Id: I4478d209c23e7468e6f36fe417acc01d7b986938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/769ff6dec9cbc9fe1c48b6493027407b37217f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/769ff6dec9cbc9fe1c48b6493027407b37217f96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
