)]}'
{"id":"openstack%2Fpuppet-oslo~791628","triplet_id":"openstack%2Fpuppet-oslo~master~I93ee3a37137232604aaa1532f4d46bf28264b677","project":"openstack/puppet-oslo","branch":"master","topic":"coordination","hashtags":[],"change_id":"I93ee3a37137232604aaa1532f4d46bf28264b677","subject":"Add the base implementation for coordination parameters","status":"MERGED","created":"2021-05-17 02:03:28.000000000","updated":"2021-05-21 10:41:42.000000000","submitted":"2021-05-21 10:39:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791628-coordination","meta_rev_id":"7a5b9b323ff5b3db909b7e29a2ef507403f6184d","_number":791628,"virtual_id_number":791628,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 10:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-05-20 11:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-05-20 19:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-05-20 19:26:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 03:57:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 11:21:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-20 19:26:04.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"daee7f493b0b0a5379a4366edc26a61164814fe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-17 02:03:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cac1dcde18bb5e76fd68d44983365d81c0ca35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 03:57:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a76979e1ee8b4441b335d4c4327196f9 : FAILURE in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/930af48a13a342f1ac2baa7dbd071d3a : FAILURE in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16cbf96061f6439fb32c142c74930bf2 : FAILURE in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbff87aea445467db603d34b1d746558 : FAILURE in 10m 38s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/df53e7f4b0ba4222bee7d90b7f5e1701 : FAILURE in 9m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42245c2c963a4566992cb378efc61785 : FAILURE in 6m 57s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1bca3fbbe61427bb0ae155e964eca91 : FAILURE in 7m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b56c9c8365344ff9ab6ec53ca6c10191 : FAILURE in 6m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0321ca8fb60540d89bf8d46b48806aa1 : FAILURE in 35m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/610839559c9c46fa9fd015f3b8550bc0 : FAILURE in 38m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94ca78521c034d0699b4686c52f6e1bd : FAILURE in 28m 45s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2b4cbf64abd452dabcd35757feb097a : FAILURE in 28m 44s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95f67e9777c747d4bc2a803072578a88 : SUCCESS in 22m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7505b016d55f4b449243f587818bdd97 : SUCCESS in 1h 27m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c527cba42b7f475090c90037871a55de : SUCCESS in 1h 37m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/80aa8bb0681e47aa9d636ee424f4a9a2 : SUCCESS in 1h 25m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b175263fae34f6eb23c1a4b8b3a262b : SUCCESS in 59m 07s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/46d1e7984b4649bc96598fb9ce21c139 : SUCCESS in 22m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/acf62fd7897f4edf92aa3654815d0369 : SUCCESS in 1h 16m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e217a6b103742bb9c5f7285f7f99468 : SUCCESS in 57m 12s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/515300427be643f1a124db6075703ff4 : SUCCESS in 1h 00m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b43a772f0c8f4cecad3aad05c0cca3f9 : SUCCESS in 1h 04m 03s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c537e68cd04841f3bb7b02012fd97ecd : SUCCESS in 1h 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9069afde98b380bb87f6a53f3f068b1077f5443c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-17 11:51:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"712f695346cafe78e00bad80247178b8a829ce33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-17 12:52:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"68c6813543bf0f86d518f9927b6c93d27b1af015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:34:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6be02d587ff4429695c610eaef308bab : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4618bd8e931453db8f7f7e9c8e84d3e : FAILURE in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23cc60e8529f4ceba9fa00270e4834a0 : FAILURE in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e88d1df75204016bcfed3e0babab8ef : FAILURE in 10m 12s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6a6e469c93145bca3d1ff1ac32863b3 : FAILURE in 12m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a0ce16d20d0446cb59312c27f55ff07 : FAILURE in 7m 41s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2901eabe57244454828738648db48825 : FAILURE in 8m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd154829f92b4cdab4cba352b597d3d4 : FAILURE in 12m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad2a17f10cf342488b8daa96d55aa276 : FAILURE in 55m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41a65a95f9074a52a50926dea720199b : FAILURE in 38m 07s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bc0dbcd76094b449b8bd3b6277bb2c9 : FAILURE in 24m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f398324c07e9403985e919be2129e174 : FAILURE in 29m 26s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a667403f4f1432aaefa1c1acf341034 : SUCCESS in 36m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/45141a0d2ec2410a848f90533ea27a25 : SUCCESS in 1h 34m 45s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/00b5853de3bb414dab09466a604e08b8 : SUCCESS in 1h 11m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/97debcacae214a0e9e7fac2ea205d169 : SUCCESS in 1h 03m 20s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fecb8fedbe343b7ab400587623e47b7 : SUCCESS in 54m 54s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff5828c0ae704899a016b5890a00476a : SUCCESS in 44m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a252c3d9cf4b464096be85430be8ff0e : SUCCESS in 1h 08m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/228b9d50c2f8425190b01bcf0a01f605 : SUCCESS in 1h 04m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ae3237ba4d84d55ba5278f8901d5c7c : SUCCESS in 1h 10m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4bcbc56637f4b349ea8e9df1044c96e : SUCCESS in 55m 15s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a17c0cb640d248b38375207d22267c7e : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":3},{"id":"a8977d44a717a7307234cb7d7c5afcff1f35ee8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 00:59:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"68286f58815dedeb710b370abe48c5e4957b876b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 01:04:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"970782075dad55385e61bee2e2383ee904caad42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 01:53:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1fb2b3a02c6787001aa686ea1477062f86e8324e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 03:34:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56bfcc51322641a6884c28c99a6d18e2 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c203abe24748409885ca025af5216e5b : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a42b8121959048e49d28e54dfad9148e : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/27ad3558586449a9ad39a2cf47336248 : FAILURE in 8m 41s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4c64406214441099aabb66aae5f5712 : FAILURE in 8m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b02fe6fdf1124ccfb01d42c073597060 : FAILURE in 6m 15s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90cbbe4300f545fbb1bbd4c45d3ffd8b : FAILURE in 6m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b097ff9b4d2e403fa2428d270dc897cf : FAILURE in 6m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bc6efee6dc646f2879c400291d9f4c3 : FAILURE in 30m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b1ad470be7d416fb9d0747e16e71360 : FAILURE in 42m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f36da07124194a939a5b15a7f523445e : FAILURE in 25m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04cfe3644ab94fa5b72a2b1823bafa86 : FAILURE in 28m 11s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef0bc04d7ae344a4aaa0f743ac15eeb4 : SUCCESS in 17m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e919fbd62f24a56931929ecb05dc401 : SUCCESS in 1h 17m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27a3afae5e8f4472a6b435f554e46967 : SUCCESS in 1h 22m 11s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3291310917c44c62bc0a878d42c75fdf : SUCCESS in 1h 12m 27s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/591f7448b1a84f3b85828d28910697a6 : SUCCESS in 1h 03m 37s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e04dbe4e6a04a8686625d471ccc41d1 : SUCCESS in 19m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e8846c1498343f4bcadaa0142392cc2 : SUCCESS in 1h 22m 22s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d38d38a9c2fc420397ee999912365485 : SUCCESS in 1h 19m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c5b3a69c8c3438f951d7396b48babfd : SUCCESS in 1h 09m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15652a1a0cd94531b4672ca0f23d54e3 : SUCCESS in 54m 19s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4e5e6def83e84ebdafe826abfd510dc3 : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":6},{"id":"759fd8cec96805029aa96828d31fe03642b0e9c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 03:36:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c23a772ccfe3b01b9df5fbb657c37432f94d6aad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 04:18:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1655cec19a93f7760ceff8111d3c2ea59dfb4910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 06:03:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cfe20044faf4353beb5c46744e8ad50 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed06cb76ce3949349f604b390cf3c4fd : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92bc203d482e4ae982a1012b1b57b747 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a865cf5bafa843d987a11ac7dc25011e : FAILURE in 9m 19s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc701b2331fe45228d3981c02c966697 : FAILURE in 13m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaff24ad5d10402fb937cca70f668005 : FAILURE in 6m 49s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0a208f53921440689fcf1d558511b9c : FAILURE in 7m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b67d863fc61547f6800d8adbb84906b2 : FAILURE in 6m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a112902f36a447198c5496c835cdd4a : FAILURE in 40m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc7853d2376449daa58ab518c92873d6 : FAILURE in 41m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/406452a0cb9c4786a0ec162a6adaddf3 : FAILURE in 28m 22s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cd84ce0590c43819be84e500cdbbd7d : FAILURE in 25m 18s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1ab4f152fda422c9bcce913771d8aa8 : SUCCESS in 21m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8e15d9836b845548a2cf8893983d10a : SUCCESS in 1h 21m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/144dcd9173a543d0ac89e6f9e87b3d99 : SUCCESS in 1h 35m 34s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95717a8ea9b3466f94d7ea3f4b08ec64 : SUCCESS in 1h 20m 58s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/40ea6a49d3ec457ea0d5e8ef75997f8b : SUCCESS in 1h 01m 21s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/81e30f47449942899edb0040a89c5afb : SUCCESS in 21m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff344b722ebf41c5bdb3fb721aedbe87 : SUCCESS in 1h 03m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3ddec867d0d40d39bc4112ede825bcb : SUCCESS in 1h 39m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3627373bfa714c97b8b17a82bd785a8b : SUCCESS in 1h 22m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d5a948072824a6a8b9ce21bebdbd0b3 : SUCCESS in 1h 01m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cafda4de875e41beae04aa837653fadf : SUCCESS in 3m 52s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4a5cb017a536465892656c24eb589665 : SUCCESS in 1h 35m 36s","accounts_in_message":[],"_revision_number":8},{"id":"4a6cf6ac18c456e29aca604ca0fe18e8dd8ecbe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 06:32:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"093634f72fafeef4cd3578e7080ed591d18d1064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 08:14:10.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c2b082211d44927a228f72b7374dece : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aca012cac4a4b25861d31794476e489 : SUCCESS in 7m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f50f130bbdde45728c6878963eecef5a : SUCCESS in 5m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/12284f756d564a22a1a5e9bdf39f8c0b : FAILURE in 12m 52s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/73aa3c573bbf4941b09df621c3671b8e : FAILURE in 15m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41dc3ec4bef74246a8c37f58cdaf52cc : FAILURE in 8m 35s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e4747ef598842ae9e952e5ef1a407e3 : FAILURE in 8m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebf8faca2a0d4261ab19502386732d8d : FAILURE in 8m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eca4e54a3d84dea9db1ffce39d9e63a : FAILURE in 49m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dd3dbbf0bdd4bc6aead139cabf5b2a6 : FAILURE in 36m 54s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7e15586852e41c1b7d4de2129fbb448 : FAILURE in 33m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85ae4c5347664f56b15cbb123fc3581e : FAILURE in 28m 54s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/924860afe4f24428bcd5604f119e00f5 : SUCCESS in 22m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/121ee0d925b844e58854ba559737d486 : SUCCESS in 1h 11m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7916749cbf44fa7aba695b2a6c97afd : SUCCESS in 1h 36m 37s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/025cf0e964c24e378372005bd89652e7 : SUCCESS in 1h 25m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7df9040d91e44a538eeed1e19f562465 : SUCCESS in 1h 04m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/32153696e2fa4116b9c9d5cad981b7d0 : SUCCESS in 25m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3570e37ed454dbb9509a78be697d351 : SUCCESS in 1h 30m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd641f49ea6642208903cbd778972f43 : SUCCESS in 1h 33m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9af7ad8631d4b05a435e9e8783c8253 : SUCCESS in 1h 08m 53s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9723569b149644a4bd9a552a272bbf6a : SUCCESS in 55m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/187af3bcbede4462952b7301aca50033 : SUCCESS in 5m 50s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7c9248e88e2a45ccb98c9b72fccb4ad7 : SUCCESS in 1h 37m 11s","accounts_in_message":[],"_revision_number":9},{"id":"ea08a5bb9ee44b9ae3b136beb73772c9923fb6c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 08:40:15.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"263417dc794346a2c2a203ce24fbe7706fd94790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 09:14:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a7bd473ca3ab98677b688f5ad1bc76db754f024d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:00:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ad58b104a6d4a9b8f98aa8461ba3f0d : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b019c8a2903f46f8ba059cd86b5b8f92 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/168e48bf8e49433bb74a3e60012a43fb : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bccc2f80c7964af899679e8f575dbd9f : SUCCESS in 9m 44s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/aece05ab39274756a0b2f5762e8b804a : SUCCESS in 9m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f6c8dfcc13b46f39e8b59635ddc34aa : SUCCESS in 6m 53s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6dad63452c0432fbad523630fcb216e : SUCCESS in 7m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f5d7ccad3ca4222b60828440c1bca25 : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b961f7df5ca429aa537ac918d9a8688 : FAILURE in 39m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a933c8701644d4ea60afaaead568684 : FAILURE in 39m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9c73663a7b24ac1a62305bd2caad9d6 : FAILURE in 28m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3c22ad128d94c1e8c0d6e5073ad1215 : FAILURE in 27m 29s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/135d0a0dcd0f40a0a2ef6d93377b448b : SUCCESS in 21m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6dc48ccd6344670abf662e0588b1d51 : SUCCESS in 1h 11m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/441480a94b9549928aee20cc6e19a2fd : SUCCESS in 1h 24m 09s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/292639172ebd493fa7184f2afb7c1d1b : SUCCESS in 1h 02m 53s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d66aee48d7874543a035789b02cf6864 : SUCCESS in 1h 01m 15s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/50b93e2bf1654bceb7878e6a2593964a : SUCCESS in 21m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/831e8dc597594707849a45ebfbc45c99 : SUCCESS in 1h 17m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bea5d00d96aa4577ba38a4cd02e9b6e9 : SUCCESS in 1h 16m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2466678505504062909989f9e74b1424 : SUCCESS in 1h 15m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a07cc54a4ce477fba5293f18553a4ce : SUCCESS in 49m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cbbd64fc9f249b58d98104fd659a23f : SUCCESS in 5m 05s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ee29028efb5a4980b9700e6b6554de70 : SUCCESS in 1h 42m 43s","accounts_in_message":[],"_revision_number":11},{"id":"cbb95fb46e4c0cda7d87bc1bf215d3b430a4f3c9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-20 11:21:14.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"7e43a5140b10e2aba835c17c6a1177f2fd754826","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-20 19:26:04.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"b438da13f285e0db908239595344984bc7e99fe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 05:02:04.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"62ddb9d94c72a6267920595e78b4e175b6302ead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:39:17.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0584111fe274a76a0bbbb975b4d72cf : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921ff75f809d4a3ebea5d0c4e6f467cf : SUCCESS in 4m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64b50a2b2aa94a84aa02230fa23086a6 : SUCCESS in 4m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/62be4fee7ec14977956160b0351fee2e : SUCCESS in 15m 05s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c8c226f434441f38c4ddc95bec58fbe : SUCCESS in 13m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61fe7802b4e442d79f12fa8a4191109b : SUCCESS in 13m 29s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/085b4ba2e6954c179986d3b254755798 : SUCCESS in 9m 31s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cfbb7d0aaf64461b006e4c3648e1aab : SUCCESS in 1h 21m 49s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27f9ee06d5274dd9afd1b9a9787a5de2 : SUCCESS in 1h 14m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a1a5c620bcf46c08e7bfc73bb59331b : SUCCESS in 1h 17m 27s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9e1400151214964b3cc8abd69036558 : SUCCESS in 1h 02m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1cdcbd72872401cb2697a391be3ba21 : SUCCESS in 6m 22s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f9e8b7fe4ee8492590f628f24c0e2f3a : SUCCESS in 1h 38m 48s","accounts_in_message":[],"_revision_number":11},{"id":"43a97f4a11a447a7161fb19e2a896075a458733c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:39:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"7a5b9b323ff5b3db909b7e29a2ef507403f6184d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:41:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47b771cd2e0d492681e61d38c285a48c : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"475d6e8bbd3429729a54ff62999de28a22da59b1","revisions":{"c4848b9a6d12ef191ed835d6d726b9ec4c7dddae":{"kind":"REWORK","_number":1,"created":"2021-05-17 02:03:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/1"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"subject":"WIP: Add basic class to manage coordination parameters and packages","message":"WIP: Add basic class to manage coordination parameters and packages\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4848b9a6d12ef191ed835d6d726b9ec4c7dddae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4848b9a6d12ef191ed835d6d726b9ec4c7dddae"}]},"branch":"refs/heads/master"},"cf883ff9d94b146af0973543645bbf90a548069b":{"kind":"REWORK","_number":2,"created":"2021-05-17 11:51:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/2"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 11:50:51.000000000","tz":540},"subject":"Add the base class for coordination","message":"Add the base class for coordination\n\nThis change introduces the oslo::coordination class, which can be used\nto manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/cf883ff9d94b146af0973543645bbf90a548069b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/cf883ff9d94b146af0973543645bbf90a548069b"}]},"branch":"refs/heads/master"},"fc42b3281eafb62cd564a6467f4e39d3565b86de":{"kind":"REWORK","_number":3,"created":"2021-05-17 12:52:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/3"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 12:52:49.000000000","tz":540},"subject":"Add the base class for coordination","message":"Add the base class for coordination\n\nThis change introduces the oslo::coordination class, which can be used\nto manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fc42b3281eafb62cd564a6467f4e39d3565b86de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/fc42b3281eafb62cd564a6467f4e39d3565b86de"}]},"branch":"refs/heads/master"},"9a8c3ce561a26f5a43d7259d020ce59f4d42610b":{"kind":"REWORK","_number":4,"created":"2021-05-18 00:59:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/4"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 00:59:43.000000000","tz":540},"subject":"Add the base class for coordination","message":"Add the base class for coordination\n\nThis change introduces the oslo::coordination class, which can be used\nto manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9a8c3ce561a26f5a43d7259d020ce59f4d42610b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/9a8c3ce561a26f5a43d7259d020ce59f4d42610b"}]},"branch":"refs/heads/master"},"d9032d5648502b560a6c629190f78511c3fb7b45":{"kind":"REWORK","_number":5,"created":"2021-05-18 01:04:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/5"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 01:04:43.000000000","tz":540},"subject":"Add the base class for coordination","message":"Add the base class for coordination\n\nThis change introduces the oslo::coordination class, which can be used\nto manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d9032d5648502b560a6c629190f78511c3fb7b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d9032d5648502b560a6c629190f78511c3fb7b45"}]},"branch":"refs/heads/master"},"6618b0e8b43d6bbde75458f573340511253e0d8b":{"kind":"REWORK","_number":6,"created":"2021-05-18 01:53:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/6"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 01:53:52.000000000","tz":540},"subject":"Add the base class for coordination","message":"Add the base class for coordination\n\nThis change introduces the oslo::coordination class, which can be used\nto manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6618b0e8b43d6bbde75458f573340511253e0d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6618b0e8b43d6bbde75458f573340511253e0d8b"}]},"branch":"refs/heads/master"},"4b341fba1dc71a88da72657265855ab4a1d9b0b8":{"kind":"REWORK","_number":7,"created":"2021-05-18 03:36:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/7"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 03:36:19.000000000","tz":540},"subject":"Add the base class for coordination","message":"Add the base class for coordination\n\nThis change introduces the oslo::coordination class, which can be used\nto manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4b341fba1dc71a88da72657265855ab4a1d9b0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/4b341fba1dc71a88da72657265855ab4a1d9b0b8"}]},"branch":"refs/heads/master"},"87b33f56eb63f56ad5bce1ee3a7e85ddb1b66d91":{"kind":"REWORK","_number":8,"created":"2021-05-18 04:18:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/8"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 04:17:01.000000000","tz":540},"subject":"Add the base implementation for coordination parameters","message":"Add the base implementation for coordination parameters\n\nThis change introduces the oslo::coordination resource type, which can\nbe used to manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/87b33f56eb63f56ad5bce1ee3a7e85ddb1b66d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/87b33f56eb63f56ad5bce1ee3a7e85ddb1b66d91"}]},"branch":"refs/heads/master"},"b5591d6fa60215199b4b89b1f87815a40b92b02c":{"kind":"REWORK","_number":9,"created":"2021-05-18 06:32:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/9"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 06:32:10.000000000","tz":540},"subject":"Add the base implementation for coordination parameters","message":"Add the base implementation for coordination parameters\n\nThis change introduces the oslo::coordination resource type, which can\nbe used to manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b5591d6fa60215199b4b89b1f87815a40b92b02c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b5591d6fa60215199b4b89b1f87815a40b92b02c"}]},"branch":"refs/heads/master"},"15c2b8e00a018bd819bebf57d1e9a11996c6771c":{"kind":"REWORK","_number":10,"created":"2021-05-18 08:40:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/10"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 08:40:07.000000000","tz":540},"subject":"Add the base implementation for coordination parameters","message":"Add the base implementation for coordination parameters\n\nThis change introduces the oslo::coordination resource type, which can\nbe used to manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/15c2b8e00a018bd819bebf57d1e9a11996c6771c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/15c2b8e00a018bd819bebf57d1e9a11996c6771c"}]},"branch":"refs/heads/master"},"475d6e8bbd3429729a54ff62999de28a22da59b1":{"kind":"REWORK","_number":11,"created":"2021-05-18 09:14:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/791628/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/28/791628/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/28/791628/11"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-17 02:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-18 09:13:55.000000000","tz":540},"subject":"Add the base implementation for coordination parameters","message":"Add the base implementation for coordination parameters\n\nThis change introduces the oslo::coordination resource type, which can\nbe used to manage paramaeters and packages used for coordination(tooz).\n\nChange-Id: I93ee3a37137232604aaa1532f4d46bf28264b677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/475d6e8bbd3429729a54ff62999de28a22da59b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/475d6e8bbd3429729a54ff62999de28a22da59b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
