)]}'
{"id":"openstack%2Fpuppet-oslo~792569","triplet_id":"openstack%2Fpuppet-oslo~master~Id2e98924bf27efa0fae98519b9b17940644e95d7","project":"openstack/puppet-oslo","branch":"master","topic":"pyvers-removal","hashtags":[],"change_id":"Id2e98924bf27efa0fae98519b9b17940644e95d7","subject":"Get rid of the $pyvers variable","status":"MERGED","created":"2021-05-21 10:13:40.000000000","updated":"2021-06-17 12:07:43.000000000","submitted":"2021-06-17 12:07:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792569-pyvers-removal","meta_rev_id":"0e1244d459f5980f8a4cb0f60dd953bf46a0e3be","_number":792569,"virtual_id_number":792569,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 12:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-16 15:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-16 17:58:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-16 17:58:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 12:08:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-12 13:41:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-16 17:58:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"9929867d004f4302488853ca6a767d8db1444985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-05-21 10:13:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f83cd7e2c287881bb8070588a36c296239c1e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:08:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9df730ff363442a3936f14f31786b044 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fec4338a335b4eff8d9af0e5be24dc73 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daac440ddeab4103908324eb0e687d07 : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/26f22d10d0984ff1a67e7339391236b3 : SUCCESS in 8m 53s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a99264b1236946868190f2a500ce6a38 : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b9274b8e1854945b9bd0a9b1ac2fd97 : SUCCESS in 11m 56s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08474bb36dbd49dc9ab001454d6b2a23 : SUCCESS in 10m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3a1439adc984f86ba99b24d120314ad : SUCCESS in 9m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d3f090e1caa455f81ea57dbe13b302b : FAILURE in 38m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7741e5e38e3b4eb297af3c2af40071a6 : FAILURE in 38m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b925f39eeb724020a0769bf344587d49 : FAILURE in 27m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47f86bd67d594a2d9a545fecfb0cacb6 : FAILURE in 30m 09s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/728de52cf7dd41cfb106d32bade41e65 : SUCCESS in 21m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d590e1ccb984d21880508bad0b1c020 : SUCCESS in 1h 15m 45s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eeacdcdd68dd4a48a8fefa5eea1aef62 : SUCCESS in 1h 44m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/73d7cb23d729461b8eb0a3d3e8875f37 : SUCCESS in 1h 06m 21s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/01dfca2f2f924ff58fe9399c2e303a9e : SUCCESS in 58m 27s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1df4f63b83f24b60b1b5fcb06e7beff8 : SUCCESS in 19m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b42b1668bdec4f30baa741653a09b8f0 : SUCCESS in 1h 08m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/593592040b0146beb685312feaa922aa : SUCCESS in 1h 16m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/633d37ac6e204215af3e645a9b60b44e : SUCCESS in 1h 04m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a94d3173150d4cec9739958a3d4e2a66 : SUCCESS in 1h 00m 13s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7037fce2dd3945a184efa6b28a27236c : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b4378a831005616303559d1dec8e578fdb92c962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-05-27 12:22:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5534b36dfdfc4a238c1a239b10a3fb3e05f2656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:43:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a05049fa39e44c1d90fa4054faed6cd9 : FAILURE in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed4faae5d4624eb1b51d6d72b0a11c5b : SUCCESS in 4m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f074b468889436a934da924ff55b76c : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/16b3cdbbf94c4c399223ce83ce6e693c : SUCCESS in 13m 27s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2eb86bc058bd4b89b6f51212c0ea78d2 : SUCCESS in 10m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b89fad46124449eaab00eee929417b13 : SUCCESS in 7m 35s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08ccf46f01534b848d6c4757d48cb7ec : SUCCESS in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e8e4632658e42b1ae663241b56ae25d : SUCCESS in 10m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12c0c1f1a4e94770bdf80e672394f975 : FAILURE in 45m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12b5746bb5d148b282f1df8e50014252 : FAILURE in 42m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7645fa45d9a8454cbd7d3f19958b7d34 : FAILURE in 42m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1dc59c75afe4e909bd87280e4db36ad : FAILURE in 32m 38s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3586a7ce732348869e4ffe885b545678 : SUCCESS in 21m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/712ec024cf5c46719bcef8d76695ddbd : TIMED_OUT in 2h 05m 54s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d04cb67b9ad3410b814cb426f6eafb3e : SUCCESS in 1h 45m 52s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1fc47f3b58345329a2f8344ac963106 : FAILURE in 1h 05m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbf0466457594c89a84ae897715551ed : FAILURE in 1h 47m 08s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad301fdf1ddc4606a9d172298bffd6d8 : SUCCESS in 25m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f01f7e1eaa0c42fa9ae43b89676a915c : SUCCESS in 1h 27m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb9299c42f054dcfa67c4c6c632588fc : SUCCESS in 1h 20m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ceaf575aeced42619b44ff7a107c79b3 : SUCCESS in 1h 15m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d20b2cbed658490babfbcfdb485740a3 : SUCCESS in 1h 08m 02s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f419644431124c4180e8bb6f9d6f205a : SUCCESS in 2h 14m 49s","accounts_in_message":[],"_revision_number":2},{"id":"2cfac2578db9431f45c0c5979ec121ee24b11306","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-09 11:55:02.000000000","message":"Topic pyvers removed","accounts_in_message":[],"_revision_number":2},{"id":"9efd1cda8f77033420d28737b11048637219ddb3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-09 11:55:05.000000000","message":"Topic set to pyvers-removal","accounts_in_message":[],"_revision_number":2},{"id":"63e8e156fc853f07cb971c84002be5c8876537f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-09 14:24:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"79acb170810c22e1e63e36db97e8304c9f81d7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:26:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6ee0db39bae4e4bad2f7fd21b740bdd : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7078e195408f44e8b2391b305c3bf851 : SUCCESS in 7m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b56137437904c62a3cce5da3345148a : SUCCESS in 4m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a32e1573fa3341768b8c5f07cc7926b3 : FAILURE in 12m 19s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/33b5c410e51740b5bcbac864039069b3 : FAILURE in 10m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ddd7ef0527e4273a80456a6774113b5 : FAILURE in 8m 52s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5065c13b7914250a8dfef64e95ec41e : FAILURE in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/782e53ef2bd24f4c9c2460b3172b5005 : FAILURE in 8m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3de3c6d8053b416fae69b6e5cdedabb1 : FAILURE in 43m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6e455effd444235bc097d726d99af0e : FAILURE in 43m 16s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aeec6ffc63245268b2ce9e3da1f1fec : FAILURE in 45m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9181050c27f41219407ef07b0c1693e : FAILURE in 28m 42s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/948a387b4e024cd491607047b04135d8 : SUCCESS in 31m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec25e5fdaee04375839118c2543658c5 : SUCCESS in 1h 16m 04s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ea6396c41f548caa7aff38eef15db33 : SUCCESS in 1h 32m 23s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f7a458ff36f4494b9d9537e3ab95806 : SUCCESS in 1h 24m 39s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cab22fc73f34528ba23611d086d78b0 : SUCCESS in 1h 03m 13s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8b112fcef5241dbbdba85d69bc5432c : SUCCESS in 35m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fba5a5454594dcca78296311cad4c2e : SUCCESS in 1h 08m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d1d47e5a9d646f696cd0ecc21b9ba4a : SUCCESS in 1h 30m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f4e791053254dbc94f128be8d96dddc : SUCCESS in 1h 17m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c1ae3085d544602b3aabf8f79e4f93d : SUCCESS in 55m 29s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/da84e76037204e77a46c62cf1a66f2db : SUCCESS in 2h 01m 19s","accounts_in_message":[],"_revision_number":3},{"id":"6c1ea5dab3c55bfa811b5a8e6c57e954bac8df03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-10 07:59:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1a58632e08876a867b7a4e05b92540e0197c22aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 10:20:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/055515c2ffe546b18461dda7ee6e3097 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a18a20d49b49de9b10ed6b61fee022 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d897c14308aa478799ced3038ac4d30d : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c3b956255754a2691d3334e5cb460b5 : FAILURE in 9m 56s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/23aadb2b63a34bc5bbb75ac0f403e070 : FAILURE in 10m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bb095c17d9e459f8b0603e65dc312bb : FAILURE in 8m 28s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f4ffcea066048f8b28f30d1d409ada9 : FAILURE in 8m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5636459ec4584c3d876e6e54b82f89ea : FAILURE in 7m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbfd03ea662e4560a7f2ae7ab92917da : FAILURE in 41m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e43ef9e4803a463aa156c33c9d4c7e04 : FAILURE in 38m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aebbe3359a0a4f3b90bd549da5001440 : FAILURE in 28m 32s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d7676ec772746b3abaaac5823f96266 : FAILURE in 30m 42s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1064c3c14a7c4407921310c24340fbba : SUCCESS in 30m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/66556bcbe6484c72b2ab5f6360595f38 : SUCCESS in 1h 11m 56s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7da9a3cfce994dc090b9cdcf3c18f4be : SUCCESS in 1h 32m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa97e44c073c4af282510ae9f9d5f42a : SUCCESS in 1h 19m 51s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4262e1eec5e646f680d11062d721ee65 : SUCCESS in 1h 04m 29s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ccc925529c849ce8f1291e93f038651 : SUCCESS in 25m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2094c2a8e694fc0858e42abdd7ebe03 : SUCCESS in 1h 12m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c67aa11e218a47be90a473c2f327cfbc : SUCCESS in 1h 38m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c63ebf3e7e04b759c0b1a7a02d1482a : SUCCESS in 1h 19m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc58653ddb384b629b89953bf5989789 : SUCCESS in 59m 29s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1273082b6275409a8909be312ab30692 : SUCCESS in 2h 15m 34s","accounts_in_message":[],"_revision_number":4},{"id":"ede2cbd3f7253cceefc98a6ea17be480acdaa1de","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-12 13:37:07.000000000","message":"Topic pyvers removed","accounts_in_message":[],"_revision_number":4},{"id":"4d72ba6aac6fca09eea57624f14d6831e9a2ee48","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-12 13:37:12.000000000","message":"Topic set to pyvers-removal","accounts_in_message":[],"_revision_number":4},{"id":"101b0510d1c07b0808566a025f9511091c1073ba","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-12 13:41:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb9d6c0abf625f1099af6d64e5c59985e8129a02","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-14 07:18:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82c139b091b0373373f9f4192cbca471c29a1566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-14 07:19:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a97045b4539f14d035c902efcda2a5aa4b8bf126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 09:30:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76211594894d482099057b45de0525a9 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a957cea2beca446289a47596978819ad : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c906f067818445d97bab8ace96177f4 : SUCCESS in 4m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c98fdeb96d94bde87188dc8c50dbd6a : FAILURE in 11m 50s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d75eee1a49ad430a91a175716b89f20d : FAILURE in 12m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/044e056fdd704322b98244a61d9d7774 : FAILURE in 7m 17s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1875f401817247879e5e9813a20c99f4 : FAILURE in 7m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b60c649523884a47b4f6d2200b513763 : FAILURE in 7m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e90997b071844cbb0a93803cf544b09 : FAILURE in 41m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80bc313f0f704ec6b6fc69b65279e8ee : FAILURE in 39m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c661d6b99e34cdd9828e27e1f22ce76 : FAILURE in 27m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85ebe118f7dc41aa846cdc0c9c9b0361 : FAILURE in 29m 50s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/76124454f2c6479fb199bcf8d82aaa03 : SUCCESS in 24m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa1203b65c154f26bd5a9a95d473f5fb : SUCCESS in 1h 16m 04s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/55c9bd1c760c4d48ab18b7ea46fd9f29 : SUCCESS in 1h 19m 52s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33228fec289d4f3ca90c8f26b61f964c : FAILURE in 37m 19s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6293ec99dd28405c996d0d72bd6651ed : FAILURE in 36m 16s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f261f75ec0664360b7de37b09459c098 : FAILURE in 40m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f43851642d094ad4bdef26adb2087925 : SUCCESS in 1h 15m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e629bb5dd64a441c81530e9effb0b6fe : SUCCESS in 1h 19m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/82e69621df04485e9421c7e6e7dbefc6 : SUCCESS in 1h 25m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a0663e265474b3d93c56b7e2c1a9297 : SUCCESS in 1h 08m 08s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/de5631760ce747a1a4920c5f63b055af : SUCCESS in 2h 07m 03s","accounts_in_message":[],"_revision_number":5},{"id":"4ca290ce85d9e4fb31ffa88618ac1e3fc6a11c92","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-14 11:48:55.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"481c79be10c2f4942510c2080e1f0a95cd9eeb4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 14:20:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51b7dfc1c60b4f8e8043cdeff6d7d9b6 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/beb27d2e0420456592389d81e0a47880 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86530fdc122446efae45cb123e2162ba : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8496dd477fb42ff84bca5dabf4b6335 : FAILURE in 11m 20s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9151f79a6b7e4abb95408f401ba78693 : FAILURE in 12m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8ac9c7593424db7aece6d032a521aa3 : FAILURE in 9m 06s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8b0d9cf006d4007a0ac831f192c1a66 : FAILURE in 9m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eee2b7a6d27f4b86af6c5069aa3dc440 : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/926d3b5f66e14c219ae08b6c4f2b045f : FAILURE in 51m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41ed06e86e754b66ab4c641a39d9120d : FAILURE in 47m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d82c7c0f7184b25a69c7a6b5324557f : FAILURE in 38m 00s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d23947d9afb74c8b9cd08a66f653973a : FAILURE in 38m 40s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfda7e9b6f20491294f360517bf3734a : SUCCESS in 24m 55s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/287b51fba3394f568bd6c9dab970eedf : SUCCESS in 1h 17m 01s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9e88642f91246189767cf97572dd7e3 : SUCCESS in 1h 22m 28s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8ab360369c64498868b5a51b5d9b37c : SUCCESS in 1h 21m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0616640ba6ee4d5fb558374e1b6273b9 : SUCCESS in 1h 08m 33s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd93f8fd59c842d4afcbc12c0661ef18 : SUCCESS in 25m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d668e275a8e1430a9c4eb91097851d9e : SUCCESS in 1h 31m 33s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/772945f244324337a02086aacd0a539a : SUCCESS in 1h 27m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/768df25046784c5db5b53d4532bf7e1b : SUCCESS in 1h 37m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5994c22a61c94260922543a6e914dfb4 : SUCCESS in 1h 10m 01s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9d3ca2e1ffec48aca3f94c577ce49d8a : SUCCESS in 2h 22m 30s","accounts_in_message":[],"_revision_number":5},{"id":"015e230f8d618a350d03a1ec6e542b1858b6e3c2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 07:57:44.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"500a42c5903689ae21a10f2ddc635ba522946768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:00:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0aae7b2399ef7609393bbceb7b57d8e7618495b5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:01:04.000000000","message":"Patch Set 6: Code-Review+2\n\nRemoved remaining reference to pyvers in unit tests.","accounts_in_message":[],"_revision_number":6},{"id":"d0e10b027c54f7913571fe0937925febac71e630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 10:03:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f115727807b94574a0c5ec90c8e49f35 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1b81c87b4884003aa1af741d567cec2 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/026cd0db58df479fb0d7b75af7084a96 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/20b1aa0e57f649dfa1bae923c08c5939 : FAILURE in 12m 11s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b3862d939a44161bfb54b166bb57ef2 : FAILURE in 12m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db5a056772da4c819518b695b6aea1f6 : FAILURE in 9m 07s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a17d60462a2d4c3fa4e164b43ebbfe69 : FAILURE in 10m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00ef394294de493d996baaa70edb99c8 : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51478ce427ac437d8883a10a5a2a87bb : FAILURE in 41m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bb77621ef0846bbb7d79d4c5b5ece73 : FAILURE in 1h 01m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a27fbe2d6b543cb896c5260f6533684 : FAILURE in 27m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/761b6751f5b84de8833ce2771607f5e9 : FAILURE in 28m 09s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/defd10c194494f41a13a065066caac7d : SUCCESS in 23m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c5b119f6e4b483ea4185eedc8df8511 : SUCCESS in 1h 14m 58s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a478816ad8d45fba2f0e5874d1e356e : SUCCESS in 1h 37m 02s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/15d4d40ef2934713aab6eb69042971ed : SUCCESS in 1h 21m 34s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4de456c5ae42426eac82452855355813 : SUCCESS in 1h 04m 57s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/daee19bc83e64e9cbe0a411da5817ba4 : SUCCESS in 25m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e7e895a888c40de89c77d34e29ade1e : SUCCESS in 1h 27m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/51eedaf9187642f394774d8888888d8a : SUCCESS in 1h 26m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/396aa697e6ca466d9629017b32d8a10b : SUCCESS in 1h 16m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9c6e64a6a244410b9166d0d1283ebb0 : SUCCESS in 1h 05m 33s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6b0f879684844b9abe7528a29993f58b : SUCCESS in 1h 58m 23s","accounts_in_message":[],"_revision_number":6},{"id":"849e5b7212dc27f0a6616a485b7181c97a81c108","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-15 11:14:36.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a9d80d4c4dc73242a776a38fa066820fa8a2b7b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:10:20.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9c554c8eaf04ab5929f9d19b4c8b0dd : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/404aa76b890142b796533d4bcafd96c9 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d979ff9737924951809eeb11b533e995 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e3e3ed8d655413d9eee04896a37b973 : FAILURE in 14m 29s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/497d398941fe4e1e8515b7fa36c31581 : FAILURE in 12m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0bcbab6c67544aa939a906f6c73f2b4 : FAILURE in 14m 18s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b067c6bf52e4aa3980fdd5101473bca : FAILURE in 9m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30acb0786d194529b6b41d533614b92b : FAILURE in 8m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3db6d68c3ee4f64b29442fbc7129e93 : FAILURE in 39m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8974dfab00240508495a00966637844 : FAILURE in 42m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0518513990c484b85929cef97cede0f : FAILURE in 27m 15s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a5156cc3f94d30aea37f51c07b553a : FAILURE in 30m 46s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8658b1af2a34ebb9a6c995d5ac6039f : SUCCESS in 40m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/aabb2ad848364a768a4ae1494b9cb63b : SUCCESS in 1h 12m 44s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24322c63727c46ed8bb2d4c96b030a9c : SUCCESS in 1h 28m 52s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/374a6bbe01ec437c9b772d1b0db559f4 : SUCCESS in 1h 21m 15s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/691ed58f4dbe4eeaacc4f3ad4a4a333e : SUCCESS in 1h 42m 30s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee6c97bcf04045faa8c40242a00f85ae : SUCCESS in 23m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4fcd26b8e1224b3dbf89c5eed49d1bfd : SUCCESS in 1h 16m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/46a5bae9787948799cf2402d3ea6148b : SUCCESS in 1h 26m 59s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad93cd8f4974461e9ca58971daabbc40 : FAILURE in 1h 23m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33945d22e938427c9774a4ed5e26767b : SUCCESS in 1h 01m 48s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d447a16252044ea8858cca289f79510f : FAILURE in 31m 57s","accounts_in_message":[],"_revision_number":6},{"id":"266975ebd240412b1ff9e4a0a8d6898d1adcb0b1","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-15 13:37:00.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4e3e09788573c9cc0f0ed5e955bee7a449f4da3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:47:10.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67ca23abb8714516b643acc21b64b4ae : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d6d39a8f46545b08301a870f1e4d9f0 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ed9ad132b8241b7a821e2a5b0ca9f4e : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/62956df38c2a454ab07f189a6eb45eab : FAILURE in 10m 46s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/85556a5a7f7744d882c40c3051dfa5b1 : FAILURE in 10m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8aa1e602d94b03a1df22597462c217 : FAILURE in 7m 41s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45eef0a3351944cd952200c1d76759e5 : FAILURE in 7m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d42d864a27545ad85469caf1d3ba96c : FAILURE in 7m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2ba1f2525214cbf893e65ddf240570d : FAILURE in 40m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07de723a65794251a2d7ff0d247ba6ab : FAILURE in 38m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d160571f74944fa9b3592f406c65a698 : FAILURE in 37m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efaf155d7bc5421f84f98cbd6d214b49 : FAILURE in 29m 34s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/32cb05637d2f428198a52ea3face86aa : FAILURE in 17m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3052e61b1f04a149df4cf166f6ec662 : SUCCESS in 1h 40m 58s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/493f1acf2844479e8f30554743fa9a5a : FAILURE in 17m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0769fef153f4f808ff3286c7bee8e93 : SUCCESS in 1h 23m 48s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c5ecc58a6dc44a497c2b59f059e1a3e : FAILURE in 14m 52s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fda97f40991452a91767c802c2d10bd : SUCCESS in 25m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a238d5dadb44c7094ce3c6f63b6c10d : SUCCESS in 1h 18m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4fe16fb2d4c64adba5a7b98861cad93e : SUCCESS in 1h 25m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef4c471006fd4790a96ac57977ffadcf : FAILURE in 14m 17s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3ddb6e68e7040a59bfa6f5d6664d7dd : SUCCESS in 1h 04m 29s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f11f9bf5c2e4d0f914a8e58ae4bd366 : SUCCESS in 2h 09m 02s","accounts_in_message":[],"_revision_number":6},{"id":"6d0f12091fb4f53062672bdde0980e94561671db","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 00:10:30.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f9fb9bfbd2f1ae78592441c7cb2a9cbb8e8cc345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 02:24:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bedecc93ad24ba6a5b2f76ee51b7f7c : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23c187ee0e0c4ec68a81f849afbbb29d : SUCCESS in 4m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be6b00c848e4b4ba01a4dad92c02ea2 : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a30eff05b5f34b779f2739553a998fb9 : FAILURE in 9m 02s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2edf200c898042eb99202b3d2af5f130 : FAILURE in 10m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b6590cc6e064e8e9d8319fd84cdc1ce : FAILURE in 8m 23s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b528247858f4027a3459f6cdd30f144 : FAILURE in 6m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e5e0022a13e4d55971e5709cf11b22c : FAILURE in 8m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2e3032701734b1c9e362bac877ecdcf : FAILURE in 40m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/839e9a0803d9473e87aa44a4ce871362 : FAILURE in 47m 50s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f320bd7986554f48a341d33e9144e016 : FAILURE in 30m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cd22128b3bf4ec98df3fa085d2dd1f3 : FAILURE in 28m 16s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/76604c624d08432ba67b9d24a76a8fae : SUCCESS in 24m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/63de16266e0a4ea995d7e38f0f4e9e06 : SUCCESS in 1h 09m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f338730bf6a4b88af4994141fee6aee : SUCCESS in 1h 40m 19s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a69d9a88e34545f5a61e95174222adca : SUCCESS in 1h 28m 25s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0643852c967345f3b2b8156b7e1e0baf : TIMED_OUT in 2h 04m 30s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/328addb45f254e0dae222f4958aa7b5b : FAILURE in 1h 07m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9c6ca0e0ea745289b511e515c187a3f : SUCCESS in 1h 19m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c45b2355883b483d97d8caf1013e13b2 : SUCCESS in 1h 32m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c276307a7301463694a775713a8113a5 : FAILURE in 1h 06m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/164dff1b626e423a952f2135f31183c6 : TIMED_OUT in 2h 05m 28s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/84b13d3b6e104fad88883e507a3592c0 : FAILURE in 2h 12m 08s","accounts_in_message":[],"_revision_number":6},{"id":"a0d63f36a26753d14a1183b9f4b1e2740678a3c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 13:01:23.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3d2aeac0a659f5189358075841d69f31a234ff6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 13:02:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b92af5b040903fc589a476a35b96f6668680d160","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 15:14:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/865a44a8456641e3b7371e012e51bc21 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d06484a8c8a64ecb8451b81529909bae : SUCCESS in 4m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/588afb781f7a41c98616a0216a7b2396 : SUCCESS in 3m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e396ad0533094c9a9b85e75fc0b44686 : SUCCESS in 12m 18s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/935bf1226eea46ebb293818ca30523d6 : SUCCESS in 12m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2af0634c38f4841b366b231663953af : SUCCESS in 7m 27s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/643716e449224ebeb4a0d83d66aea93e : SUCCESS in 8m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8247db1f2145949f008b22f6201eb3 : SUCCESS in 7m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7db7da66ff674906becb32b31f980f29 : FAILURE in 40m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3e9a0aadaf3439d9f60a4ef54c961f9 : FAILURE in 42m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/064072dc44cb4235acbda0df1418db16 : FAILURE in 29m 22s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64e8cf053a8c4ab583bc56c30257bbaa : FAILURE in 31m 54s (non-voting)\n- puppet-openstack-integration-6-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cd1c20a18d54256be1e612f451c5a21 : SUCCESS in 24m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f4715f8c00d4c3a9a2592fee8186f71 : SUCCESS in 1h 13m 21s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f7a7d95155f445a913cccffb6d0649a : SUCCESS in 1h 31m 48s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd858300c60c48d08effd5ad45ffeaee : SUCCESS in 1h 22m 19s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b7fbb56eb6b4d7b992c0e56d6cdc3ec : SUCCESS in 1h 04m 04s\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/30e97f96dea047edb969f45ab2c350be : SUCCESS in 24m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1016b8321d5c4234ab32178135fa7476 : SUCCESS in 1h 16m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/200d97c841bf4997a900c748a5cd32b4 : SUCCESS in 1h 25m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7e014be4cda42c68903796cd516baf5 : SUCCESS in 1h 15m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/37e1abfe068b4a33909f21c6ad827a37 : SUCCESS in 1h 03m 00s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1f6e03836ccb4a008c638de293deb60c : SUCCESS in 2h 05m 35s","accounts_in_message":[],"_revision_number":7},{"id":"ae9c6b200676d2dc426d3674a57f51cce98340f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 15:44:41.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e9144501ea9633c5ab0e4d90f6449e6dd00cb048","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-16 17:58:51.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"dc50e1f76cf06467a58afd22a30745ecfe7b18e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 17:59:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b7ad87dd8581bfea58ed5bee6605c78a5946a329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:07:43.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a494edcda9742a09439dff9c9770acb : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88eb090e684d4866825a4f0d0cf1b9ef : SUCCESS in 4m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1a7c8dc6d82468094d4c975246b9d20 : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a1fcb128b154eec885efbc01e99b18f : SUCCESS in 12m 06s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6ae00754ad14cf6afa9c9da08872547 : SUCCESS in 10m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d0af6d318f745d881ff89229d50b696 : SUCCESS in 7m 29s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb67667ed38648cabcc884c48add32b9 : SUCCESS in 8m 23s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc9000e366a844029e6fe6ddb01c7bb1 : SUCCESS in 1h 18m 06s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9544a844d5843bdbcded1278083fad4 : SUCCESS in 1h 24m 39s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd666a147c2a43d78f3082634d6b6cec : SUCCESS in 1h 19m 51s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5e22679203b429e8a87d56d9faa66b3 : SUCCESS in 57m 29s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1e0a83cf27f84b9e9c9176576a1f6950 : SUCCESS in 1h 57m 49s","accounts_in_message":[],"_revision_number":7},{"id":"0e1244d459f5980f8a4cb0f60dd953bf46a0e3be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:07:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3a51feed897f8048916ce59327ac490294fc0875","revisions":{"0edadf0717708bb8d4466ab157655427bb0581a6":{"kind":"REWORK","_number":1,"created":"2021-05-21 10:13:40.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/69/792569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/792569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/1"}}},"commit":{"parents":[{"commit":"d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d434f1b8917ac0a8a5cc0e60a6b7d4e19ce4a6b3"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: Id2e98924bf27efa0fae98519b9b17940644e95d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0edadf0717708bb8d4466ab157655427bb0581a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/0edadf0717708bb8d4466ab157655427bb0581a6"}]},"branch":"refs/heads/master"},"c4a5f43674590d8af3fadc0e151ddfec76945a36":{"kind":"REWORK","_number":2,"created":"2021-05-27 12:22:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/69/792569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/792569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/2"}}},"commit":{"parents":[{"commit":"6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-27 12:22:08.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: Id2e98924bf27efa0fae98519b9b17940644e95d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4a5f43674590d8af3fadc0e151ddfec76945a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/c4a5f43674590d8af3fadc0e151ddfec76945a36"}]},"branch":"refs/heads/master"},"5c6c908d099bb5c32e32baf0e2a1c020ca513400":{"kind":"REWORK","_number":3,"created":"2021-06-09 14:24:03.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/69/792569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/792569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/3"}}},"commit":{"parents":[{"commit":"6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-09 14:23:34.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nThis patch also removes python-sqlite2 from Debian since this package\ndoesn\u0027t exist anymore in Bullseye.\n\nChange-Id: Id2e98924bf27efa0fae98519b9b17940644e95d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5c6c908d099bb5c32e32baf0e2a1c020ca513400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/5c6c908d099bb5c32e32baf0e2a1c020ca513400"}]},"branch":"refs/heads/master"},"145d0799cf145c87d4f0080605711cc2bd729792":{"kind":"REWORK","_number":4,"created":"2021-06-10 07:59:04.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/69/792569/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/792569/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/4"}}},"commit":{"parents":[{"commit":"6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-10 07:58:54.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nThis patch also removes python-sqlite2 from Debian since this package\ndoesn\u0027t exist anymore in Bullseye.\n\nChange-Id: Id2e98924bf27efa0fae98519b9b17940644e95d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/145d0799cf145c87d4f0080605711cc2bd729792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/145d0799cf145c87d4f0080605711cc2bd729792"}]},"branch":"refs/heads/master"},"aafd4ec0e7a6aa188d53b0df5c2342a2495e641f":{"kind":"REWORK","_number":5,"created":"2021-06-14 07:19:35.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/69/792569/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/792569/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/5"}}},"commit":{"parents":[{"commit":"6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-14 07:18:16.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: Id2e98924bf27efa0fae98519b9b17940644e95d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/aafd4ec0e7a6aa188d53b0df5c2342a2495e641f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/aafd4ec0e7a6aa188d53b0df5c2342a2495e641f"}]},"branch":"refs/heads/master"},"a4a7aa0cfd4b6525a4c1713056d4ccc2a8bc833e":{"kind":"REWORK","_number":6,"created":"2021-06-15 08:00:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/792569/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/792569/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/6"}}},"commit":{"parents":[{"commit":"6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 07:59:46.000000000","tz":540},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: Id2e98924bf27efa0fae98519b9b17940644e95d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a4a7aa0cfd4b6525a4c1713056d4ccc2a8bc833e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a4a7aa0cfd4b6525a4c1713056d4ccc2a8bc833e"}]},"branch":"refs/heads/master"},"3a51feed897f8048916ce59327ac490294fc0875":{"kind":"REWORK","_number":7,"created":"2021-06-16 13:02:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/792569/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/69/792569/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/69/792569/7"}}},"commit":{"parents":[{"commit":"6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35","subject":"Merge \"TripleO: Reduce tempest test scope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6c1df38ce8b3f9e7160e0b328ba8f86e1053bd35"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:13:25.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-16 13:02:13.000000000","tz":540},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: Id2e98924bf27efa0fae98519b9b17940644e95d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3a51feed897f8048916ce59327ac490294fc0875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3a51feed897f8048916ce59327ac490294fc0875"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
