)]}'
{"id":"openstack%2Fpuppet-oslo~796373","triplet_id":"openstack%2Fpuppet-oslo~master~I98e31278591b3022c4ab2f635f7c830d9b4544e6","project":"openstack/puppet-oslo","branch":"master","topic":"drop-fedora","hashtags":[],"change_id":"I98e31278591b3022c4ab2f635f7c830d9b4544e6","subject":"Drop Fedora support","status":"MERGED","created":"2021-06-15 05:58:41.000000000","updated":"2021-06-17 12:09:05.000000000","submitted":"2021-06-17 12:07:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796373-drop-fedora","meta_rev_id":"3a63abe1e82b0650f18a394bcb6a387adff1a769","_number":796373,"virtual_id_number":796373,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 12:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-17 12:09:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-16 17:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-16 17:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 06:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 08:08:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-16 17:58:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e8d45c33902c3f64489fe72b40f45461fd04d99c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-06-15 05:58:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23558ff6ca7008d5db037e645b5f982eb537bb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 06:08:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/359c4ac2a6a8451bb28de794c671df26 : SUCCESS in 2m 56s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a237acbf4d0b4bc98abded3e6876bfee : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1804cdfed14646e2a3def982d0b92ce8 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d68e5ac3727954ad18d8cd703872eff731357197","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 07:57:55.000000000","message":"Topic set to drop-fedora","accounts_in_message":[],"_revision_number":1},{"id":"aa71f1ae921fda63558025f6df2540f74d39062f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 07:58:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0eca4475cd5423eceaa13abfa90e010db663532e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:00:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6b937bf6c92f780dff5d97252a2f683d654ea665","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 08:08:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6d9dde04c140eb9b04a0f635ef537c4c6521f6a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 08:11:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/625143e998a14ac8b261e1ee242eefad : SUCCESS in 3m 28s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09b467b0d78541ca944352038f4e9425 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08cb9628144247edb454bdbd574e3571 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"f3152d581421e57991d875210dc6b9feb6517f4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 13:03:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c5ae59c288076ea05b4c0ea452e2f1bff2101e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:19:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/3cacf164d8ad45608d9a73b3b94c418e : SUCCESS in 3m 39s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b8beada2424c4e932b057d0999d79e : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c76475d9dfa49a59ce35f57f7990652 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":4},{"id":"256fd8903ac6548875fc6ed1f37548613d73eddb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-16 17:58:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ac6303de47d491fbbf3dcef78d4599a0ba7b1b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 17:59:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"06c9f76e5cd8d8543b937f68791fa67022b789bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:07:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/3bb14456f7e64732abb1af2e15a8861a : SUCCESS in 3m 56s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff4e07788ef54960b7c5a793f624f5b7 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d134083be2741a992cb8f7a62e6a9bb : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4},{"id":"2e0b5eb0b4dee52db8da19e86e687ae08718f67a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:07:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a57d1dc0e7b7d5b9ffc80cbe032f4d0d894c84d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 12:09:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f935c37fb0d245f9bcb335761bb8dd52 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7468d28bf9119656c0e04051ace483f30ba2cf48","revisions":{"dd950fd9db771ff29f3ffa9fd507c9daa2747bba":{"kind":"REWORK","_number":1,"created":"2021-06-15 05:58:41.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/73/796373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/73/796373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/1"}}},"commit":{"parents":[{"commit":"01453353c3765d8635f06f305999fa24ced93ecc","subject":"Prepare Xena M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/01453353c3765d8635f06f305999fa24ced93ecc"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-06-15 05:57:47.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-06-15 05:58:35.000000000","tz":480},"subject":"Drop Fedora support","message":"Drop Fedora support\n\nFedora support is never tested, and has been unmaintained for a while.\nBecause we don\u0027t expect any actual user using OpenStack on Fedora, this\nchange drops support for Fedora directly.\n\nChange-Id: I98e31278591b3022c4ab2f635f7c830d9b4544e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/dd950fd9db771ff29f3ffa9fd507c9daa2747bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/dd950fd9db771ff29f3ffa9fd507c9daa2747bba"}]},"branch":"refs/heads/master"},"6b109ff8d2d5900ee28a89bac7ebef55d4c8a21a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-15 07:58:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/796373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/73/796373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/2"}}},"commit":{"parents":[{"commit":"aafd4ec0e7a6aa188d53b0df5c2342a2495e641f","subject":"Get rid of the $pyvers variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/aafd4ec0e7a6aa188d53b0df5c2342a2495e641f"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-06-15 05:57:47.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 07:58:09.000000000","tz":0},"subject":"Drop Fedora support","message":"Drop Fedora support\n\nFedora support is never tested, and has been unmaintained for a while.\nBecause we don\u0027t expect any actual user using OpenStack on Fedora, this\nchange drops support for Fedora directly.\n\nChange-Id: I98e31278591b3022c4ab2f635f7c830d9b4544e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6b109ff8d2d5900ee28a89bac7ebef55d4c8a21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/6b109ff8d2d5900ee28a89bac7ebef55d4c8a21a"}]},"branch":"refs/heads/master","description":"Rebase"},"10b58d496eac885aded60432a6dcf96bbb8be4e0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-15 08:00:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/796373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/73/796373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/3"}}},"commit":{"parents":[{"commit":"a4a7aa0cfd4b6525a4c1713056d4ccc2a8bc833e","subject":"Get rid of the $pyvers variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a4a7aa0cfd4b6525a4c1713056d4ccc2a8bc833e"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-06-15 05:57:47.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-15 08:00:15.000000000","tz":0},"subject":"Drop Fedora support","message":"Drop Fedora support\n\nFedora support is never tested, and has been unmaintained for a while.\nBecause we don\u0027t expect any actual user using OpenStack on Fedora, this\nchange drops support for Fedora directly.\n\nChange-Id: I98e31278591b3022c4ab2f635f7c830d9b4544e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/10b58d496eac885aded60432a6dcf96bbb8be4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/10b58d496eac885aded60432a6dcf96bbb8be4e0"}]},"branch":"refs/heads/master","description":"Rebase"},"7468d28bf9119656c0e04051ace483f30ba2cf48":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-16 13:03:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/796373/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/73/796373/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/73/796373/4"}}},"commit":{"parents":[{"commit":"3a51feed897f8048916ce59327ac490294fc0875","subject":"Get rid of the $pyvers variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/3a51feed897f8048916ce59327ac490294fc0875"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-06-15 05:57:47.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-16 13:03:36.000000000","tz":0},"subject":"Drop Fedora support","message":"Drop Fedora support\n\nFedora support is never tested, and has been unmaintained for a while.\nBecause we don\u0027t expect any actual user using OpenStack on Fedora, this\nchange drops support for Fedora directly.\n\nChange-Id: I98e31278591b3022c4ab2f635f7c830d9b4544e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7468d28bf9119656c0e04051ace483f30ba2cf48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/7468d28bf9119656c0e04051ace483f30ba2cf48"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
