)]}'
{"id":"openstack%2Fpuppet-oslo~796558","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fvictoria~If2df7664e43d62ceb979fe597e65b1ce046d00cc","project":"openstack/puppet-oslo","branch":"stable/victoria","topic":"tripleo-stable/wallaby-stable/victoria","hashtags":["puppet-tripleo-job"],"change_id":"If2df7664e43d62ceb979fe597e65b1ce046d00cc","subject":"Use tripleo puppet job instaed of tripleo job","status":"MERGED","created":"2021-06-16 02:16:59.000000000","updated":"2021-06-17 22:35:14.000000000","submitted":"2021-06-17 22:35:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796558-tripleo-stable/wallaby-stable/victoria","cherry_pick_of_change":786780,"cherry_pick_of_patch_set":1,"meta_rev_id":"7ab31935b33e65a52d0929a052ed6d1808cdb805","_number":796558,"virtual_id_number":796558,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 22:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-17 01:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-16 17:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-17 01:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 04:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 17:59:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"caaf0081cc7134b9dc66ae1a884a06eddf17bdbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 02:16:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"adad03c022bead03e67e65d566c729e8a8c0e79e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 02:17:19.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"ee3347ebe9a822ddf1963ee7aedacdcacab434d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 04:15:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2e23c72971b40c59aad03af6beee5a0 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b65bb2fb7df94210be6ce02b51fe6571 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/5646388a56bd48b484a1016defba0879 : FAILURE in 13m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5aa201b4d5894669a597d9d0174ff054 : SUCCESS in 10m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbb2928a95624719ac7f5bcf5bee3c8f : SUCCESS in 9m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/853df9f9654b40b9bd56fa87a171186c : SUCCESS in 1h 03m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a46f57ac88824b479b4c73893fa23838 : SUCCESS in 54m 09s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83ac20649cf645f39b968824ea12b209 : SUCCESS in 1h 01m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4fc54cdaf3b47429bafb4a40589b43c : SUCCESS in 39m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ef163a11e4454846b84b78166b41a421 : SUCCESS in 1h 16m 39s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42fd28bac2ce4772844e8399d14f96d9 : SUCCESS in 1h 28m 50s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d9cabb108414d2083eba7e3b86ac734 : SUCCESS in 1h 22m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/faea14145faa4f63a93c92278c2e7588 : SUCCESS in 1h 07m 18s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3918228b33584f849ac2700eea0e19bf : SUCCESS in 1h 11m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a8a2cfe4b3d4f2eb245beaaaf594b9e : SUCCESS in 1h 32m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd599ed4e9264237b80984a7ae476b3c : SUCCESS in 1h 31m 46s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d4a3b6ed220424bb02fa5ec8cbd73c5 : SUCCESS in 59m 46s (non-voting)\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ff0619264401422381e7b433acfe799d : SUCCESS in 1h 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9b917142126118d73bcf1bec2342152f0701d392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-16 09:09:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7583103738fd107aa738822250ce551af1250003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:31:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f113484b76cb4e74898ba833d75129e7 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6fdb5d54afd410da6715c6e26114892 : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/962ea3cdaa9f40d687ae4cd7fb7be5dd : SUCCESS in 11m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e975632953e4cf8b9d9a6ff5ca25e8e : SUCCESS in 7m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8860fbc9d33c4b248d2743af92a5abf2 : SUCCESS in 9m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99ae43e5ff474dd4bdd8d96a11ccf78b : SUCCESS in 57m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/131f7eed2a3b4a6087213544f66c27dd : SUCCESS in 48m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fc8ab4054e14a68bd3b04fe0e869c27 : SUCCESS in 48m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44d229e72ffc4086b2d3b0140a8fdc79 : SUCCESS in 47m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb472ce62e8544bab83d1ccb8ed33faa : SUCCESS in 1h 16m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/22ae5291831a42d6a02fa7ab043daed7 : SUCCESS in 1h 38m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9b2817117d0a429487fcad809edac56c : SUCCESS in 1h 22m 56s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d56b002bd261409eafce331bc626b707 : SUCCESS in 53m 12s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6adcb895285046c68b664e9aaec598a3 : SUCCESS in 1h 15m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cdbc85b34b394763ad0016ef840f8b57 : SUCCESS in 1h 42m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b8d3548af3b4849ac4b621cb7c2660e : SUCCESS in 1h 36m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8077394d83eb42ff968351760768edee : SUCCESS in 1h 06m 06s (non-voting)\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/02d68ecfa20647e2b03c4f0b503d2d50 : SUCCESS in 1h 51m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b53c82f044119f08a83a9b37c7e14c209a5b9249","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-16 17:59:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"738d159a672450e358a08bfcc658b1fc5e893e0d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-17 01:01:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI\u0027ll move this ahead now... because of this now all job for puppet-oslo is sent to the tripleo queue, which tends to result in looong job time.","accounts_in_message":[],"_revision_number":2},{"id":"3d7b8acad9d9f95bfc0d20fc4ff6d3e00980be75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 01:03:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9310cdb469c258a2296eff6f679d8c36327aa248","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:35:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/397ee2306f6f4a31a5f3d24d0a75e76f : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8a55e019e8049c982b80481061c26c9 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cb13bfa5d0446a193e762d3df576196 : SUCCESS in 10m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bb43384e67342b0b271a40376574d24 : SUCCESS in 8m 35s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a68abb68fa8b4fcdbb4470086ace474f : SUCCESS in 1h 04m 22s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/34abdb7630f24ed29d2ad1029f977dd6 : SUCCESS in 1h 23m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0c04872f37b344128dd5bbe2e1598564 : SUCCESS in 1h 30m 06s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1047f319b2634b5aa1b4b9726ee3472a : SUCCESS in 49m 39s\n- tripleo-puppet-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/eafce088734140cab3f93122ec1a5049 : SUCCESS in 1h 43m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7ab31935b33e65a52d0929a052ed6d1808cdb805","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:35:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"08d8f20db261b2d914f569dbe4db61c82ed05807","revisions":{"b28747fac1fc3c06cefb6cdaee069ce15466e91e":{"kind":"REWORK","_number":1,"created":"2021-06-16 02:16:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/796558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/58/796558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/1"}}},"commit":{"parents":[{"commit":"a0d1e4650e8be4008628dbc68ea507e09735d2c9","subject":"Add TLS options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/a0d1e4650e8be4008628dbc68ea507e09735d2c9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-01 13:21:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-16 02:16:59.000000000","tz":0},"subject":"Use tripleo puppet job instaed of tripleo job","message":"Use tripleo puppet job instaed of tripleo job\n\n... so that we define all tripleo jobs in puppet moudles based on\nthe base job definition.\n\nChange-Id: If2df7664e43d62ceb979fe597e65b1ce046d00cc\n(cherry picked from commit d2f0bcff69fd9ddbb6971f4ec40f9e412e91b1ca)\n(cherry picked from commit ae67cd33eb1f6a410aae79283951b6ceac642adc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b28747fac1fc3c06cefb6cdaee069ce15466e91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b28747fac1fc3c06cefb6cdaee069ce15466e91e"}]},"branch":"refs/heads/stable/victoria"},"08d8f20db261b2d914f569dbe4db61c82ed05807":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-16 09:09:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/796558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/58/796558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/58/796558/2"}}},"commit":{"parents":[{"commit":"d1036810846aff63df64e85a43ba9b4b6f52df47","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/d1036810846aff63df64e85a43ba9b4b6f52df47"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-01 13:21:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-16 09:09:44.000000000","tz":0},"subject":"Use tripleo puppet job instaed of tripleo job","message":"Use tripleo puppet job instaed of tripleo job\n\n... so that we define all tripleo jobs in puppet moudles based on\nthe base job definition.\n\nChange-Id: If2df7664e43d62ceb979fe597e65b1ce046d00cc\n(cherry picked from commit d2f0bcff69fd9ddbb6971f4ec40f9e412e91b1ca)\n(cherry picked from commit ae67cd33eb1f6a410aae79283951b6ceac642adc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/08d8f20db261b2d914f569dbe4db61c82ed05807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/08d8f20db261b2d914f569dbe4db61c82ed05807"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
