)]}'
{"id":"openstack%2Fpuppet-oslo~799364","triplet_id":"openstack%2Fpuppet-oslo~stable%2Ftrain~I4478d209c23e7468e6f36fe417acc01d7b986938","project":"openstack/puppet-oslo","branch":"stable/train","topic":"listopts-stable/wallaby-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"I4478d209c23e7468e6f36fe417acc01d7b986938","subject":"Cors: Accept array for ListOpts","status":"MERGED","created":"2021-07-05 03:04:12.000000000","updated":"2021-07-07 21:32:15.000000000","submitted":"2021-07-07 21:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799364-listopts-stable/wallaby-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":799363,"cherry_pick_of_patch_set":1,"meta_rev_id":"8b9e9154d25daf65114a34cc52c27cc99603d144","_number":799364,"virtual_id_number":799364,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-07 21:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-07-07 20:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-07-07 20:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-07-07 20:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 04:10:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-07 20:07:04.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-07 20:09:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"2990004b156f2cbcc32863e97cdb360471e14ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-05 03:04:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"0207055f8135b912b35b186c62e2ce5e14b869fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 04:10:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de62571c98694c3aa17c2c1ccb55de46 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/161f761bb62b4222946326071edf0824 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/038899d07627404f98be697a0c6443a1 : SUCCESS in 4m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9f59e4d3075941eab66f3f59d874baf9 : SUCCESS in 12m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/18fee74c9a1b42c49edaa4fbf71cd112 : SUCCESS in 10m 18s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/0317b3fd22994d888377da6e37673ace : SUCCESS in 9m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08556f9c8aa747e99aed8df1e33fad58 : SUCCESS in 10m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09c0c4bfd0654ffdbf36e147d1282c21 : SUCCESS in 14m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/fa2eb7ca63624dadac8bc76e1911d595 : SUCCESS in 55m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ad78bf5e8c2144ab9cc7806d061fa0e4 : SUCCESS in 1h 01m 35s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8ff7c0c2400d47d0852e636b1a430236 : SUCCESS in 59m 18s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/29366c498634484ea1a1d1de87471f60 : SUCCESS in 48m 20s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a5a0e2b613f344f88f3ea4abcdab877c : SUCCESS in 40m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fad61338f0540f0afff7e4c110386c5 : SUCCESS in 46m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f17343b541344728bd2e469027530d40 : SUCCESS in 33m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ba5e81c771c443cbbde7ada574916502 : SUCCESS in 36m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e7903153be534301a17ff7b70202ce68 : SUCCESS in 53m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7029c46246a44951930722740bdaf309 : SUCCESS in 58m 28s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5d70eace7ee04334867ee480d4331dbb : SUCCESS in 49m 42s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a1e0bf5754a144ed870af3ea912bb372 : SUCCESS in 35m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/498598343297446a858afabb42acd6f7 : SUCCESS in 9m 08s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57df5feb44814eb8afa8c86447759a0c : SUCCESS in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64ff224852eec2aeffc5d066b9614541ff4a49ce","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-07 20:07:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70aaf0ac42379c2986fdee55755e457d7ade1265","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-07-07 20:09:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"186081cd11ff215a1647229d8db4b66984385a88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 20:10:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa5783d48d10fd9bc2f3af467c1567f8f5f332b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 21:32:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40108ff57f144cbdb8a1f09d95b56af0 : SUCCESS in 5m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14b043fb36c842129393f413c7bd5f83 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77bb0779e89246bb966142985ebaaed6 : SUCCESS in 5m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0685d585fcf4480c80faa4aefe33db6e : SUCCESS in 19m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8ec2992872042ef8d3d0f0d7368aa8f : SUCCESS in 12m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/fc0603b6a1564963b9b9555317465670 : SUCCESS in 57m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7c8f1da769e440d083a7dfc8cffcf77b : SUCCESS in 1h 12m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/12dcfebbb0294b5fa53273ca154db0c1 : SUCCESS in 1h 16m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/b9c8091062f3445bac450c2fee3b53fb : SUCCESS in 52m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/81c29b81a4004bbd87339df360212dfb : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8b9e9154d25daf65114a34cc52c27cc99603d144","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 21:32:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78c768e61643f86db4031f3e03eb26ea24d61bb9","revisions":{"78c768e61643f86db4031f3e03eb26ea24d61bb9":{"kind":"REWORK","_number":1,"created":"2021-07-05 03:04:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/799364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/64/799364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/64/799364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/64/799364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/64/799364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/64/799364/1"}}},"commit":{"parents":[{"commit":"883fa530a6dc34d6d62c611a302273f9b9578a03","subject":"Prepare the final stable/train release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/883fa530a6dc34d6d62c611a302273f9b9578a03"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 15:00:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-05 03:04:12.000000000","tz":0},"subject":"Cors: Accept array for ListOpts","message":"Cors: Accept array for ListOpts\n\nThe following four parameters of cors middleware are ListOpt, thus\naccepting array for these options allows users to define parameters\nin a similar format.\n\n - allowed_origin\n - expose_headers\n - allow_methods\n - allow_headers\n\nChange-Id: I4478d209c23e7468e6f36fe417acc01d7b986938\n(cherry picked from commit 769ff6dec9cbc9fe1c48b6493027407b37217f96)\n(cherry picked from commit 83e2c42821c9f1f7cba052355b90b38b97198f95)\n(cherry picked from commit 5f6a5967d588d20f42f70ce1178726f0d29ec399)\n(cherry picked from commit be425dd82d82425ccc9852b62280d6ffe96fb9b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/78c768e61643f86db4031f3e03eb26ea24d61bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/78c768e61643f86db4031f3e03eb26ea24d61bb9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
