)]}'
{"id":"openstack%2Fpuppet-oslo~804011","triplet_id":"openstack%2Fpuppet-oslo~stable%2Fwallaby~Iafb03a6cd57859d11713d12b1c939a9fe782ee65","project":"openstack/puppet-oslo","branch":"stable/wallaby","topic":"bug/1939088-stable/wallaby","hashtags":[],"change_id":"Iafb03a6cd57859d11713d12b1c939a9fe782ee65","subject":"Replace deprecated is_string method","status":"MERGED","created":"2021-08-09 23:42:55.000000000","updated":"2021-08-10 16:29:20.000000000","submitted":"2021-08-10 16:29:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804011-bug/1939088-stable/wallaby","cherry_pick_of_change":803664,"cherry_pick_of_patch_set":2,"meta_rev_id":"4c3d2f06be9f36a615fea8c3f5a12c2c5d049ab2","_number":804011,"virtual_id_number":804011,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 16:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-08-10 06:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-08-10 14:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-08-10 14:26:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 01:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 06:43:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-08-10 14:26:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a63d4e9719f8478580fabbf2ff716c2ad0e201e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-09 23:42:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a3b60bb4a2e4404ab0f0aadcd4400a3335d95c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 01:30:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b187e25768d4b73884b866de8495d6c : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70c64fb2c3bc4466a05b33a7b8bccd3a : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/52de25d168a74a56b8e0f55766073949 : SUCCESS in 9m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea963f9999f84f32b6e5a20f8f70188b : SUCCESS in 6m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5a8df70673647819a385dfc29b95edf : SUCCESS in 6m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/727b14aaeb554ccbb874f6f79ce54302 : FAILURE in 39m 36s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85380e75649d4fd0aa1c3db2b4edaabc : FAILURE in 38m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78703eae45e74bcb957541d7b88fa9a1 : FAILURE in 27m 20s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/220c1a60b9414f80be9fa45b448fe2c0 : FAILURE in 27m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/309b773733f641bfad3cfa3b4ca2b916 : SUCCESS in 1h 20m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ac674af9da44279a93d119b449391c1 : FAILURE in 1h 28m 30s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3639c8292ba547c6a5702d4c43c57eb1 : SUCCESS in 1h 09m 51s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee9de11f9aa54534a7a1318082a8453b : SUCCESS in 59m 17s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/99cdaa4062b24571b722ee0faadc22e6 : SUCCESS in 1h 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d7f3351c747bf12889a26406fc2a80a354d9d245","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-10 03:55:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c4fb16a65394ac18aa9c4757e384584a68e5ede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 05:44:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9f4cee26ed44062b1ba9d52987e38cf : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f20b063573f45cda790db1846745f3d : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba9be8ac3f5c496fac36cdcd5ad2ce1a : SUCCESS in 10m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2976cd267709454b9c059f00a344877e : SUCCESS in 16m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79cebd94cada410db59e5ed11f3ca070 : SUCCESS in 9m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db0469581eb148bc9ee464e7d319dbcb : FAILURE in 44m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d74217d291df4009ba89f6685cab78ca : FAILURE in 44m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8fca26635da48dc894a72d018313845 : FAILURE in 29m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3625da69d8db4ed1a2588532849878c2 : FAILURE in 30m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6889d857c4ca488990c01764cb55ec4b : SUCCESS in 1h 15m 45s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e40c3d97a4f44f98319799cb20fee6d : SUCCESS in 1h 28m 31s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ef86737f4104d359da09b10ae3b0349 : SUCCESS in 1h 16m 49s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb662d118e1449f3821877701a5df20c : SUCCESS in 57m 30s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/492b48348f5c475b991a485657761bb1 : SUCCESS in 1h 46m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9dea3ff6ee6f9c11d6d7dee867ac1e7508843798","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-10 06:43:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04fc62585b38c51424e05b3cf86ae64a2bf221de","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-10 14:26:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a85dc491e9e719c374dd76fe170cfd0cec4ffe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:27:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b76877f76685c4c91614c7a7aed8efd977e906f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 16:29:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5eef343a18f4daaaa6d1a77af8f7b97 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/654e2261caa945f4b8cd1e1c19e40014 : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b262f42171240d1b583291dfd40ecc5 : SUCCESS in 12m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa75160e7aba4d2999bdc58bbd624cdf : SUCCESS in 9m 41s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcabc8e0a2874f7684a621f17c9996bb : SUCCESS in 1h 12m 24s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/946196f32ba542428d96f87d0c5ee97a : SUCCESS in 1h 18m 55s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8d83f48299245d793145e9a1d7d6569 : SUCCESS in 1h 16m 17s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b4470a6e43942f8b84d5b9a6af9ed19 : SUCCESS in 1h 05m 23s\n- puppet-oslo-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d36c611e40ed4bed8d83694836b9cabf : SUCCESS in 1h 50m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4c3d2f06be9f36a615fea8c3f5a12c2c5d049ab2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 16:29:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e9d3f46ba713fec05d806818fe2331c4e405db8","revisions":{"1e9d3f46ba713fec05d806818fe2331c4e405db8":{"kind":"REWORK","_number":1,"created":"2021-08-09 23:42:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/804011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/11/804011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/804011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/804011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/11/804011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/11/804011/1"}}},"commit":{"parents":[{"commit":"83e2c42821c9f1f7cba052355b90b38b97198f95","subject":"Cors: Accept array for ListOpts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/83e2c42821c9f1f7cba052355b90b38b97198f95"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 01:39:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-09 23:42:55.000000000","tz":0},"subject":"Replace deprecated is_string method","message":"Replace deprecated is_string method\n\n... and simplify how $::os_service_default, which is a String, is\nhandled.\n\nCloses-Bug: #1939088\nChange-Id: Iafb03a6cd57859d11713d12b1c939a9fe782ee65\n(cherry picked from commit 8f6788f10f9c5df893beda7a2cb3ecfd6ee11ac3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1e9d3f46ba713fec05d806818fe2331c4e405db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/1e9d3f46ba713fec05d806818fe2331c4e405db8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
