)]}'
{"id":"openstack%2Fpuppet-oslo~805491","triplet_id":"openstack%2Fpuppet-oslo~stable%2Ftrain~Iba93ffdacf45096a6db65f3f0dfcfaf7d87574f2","project":"openstack/puppet-oslo","branch":"stable/train","topic":"idle_timeout-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"Iba93ffdacf45096a6db65f3f0dfcfaf7d87574f2","subject":"Fix the wrong default value of idle_timeout","status":"MERGED","created":"2021-08-21 18:40:50.000000000","updated":"2021-08-24 08:59:27.000000000","submitted":"2021-08-24 08:59:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805491-idle_timeout-stable/victoria-stable/ussuri-stable/train","meta_rev_id":"3dcf303ac3c5dfcd7be8e19dc874873d8f68c72f","_number":805491,"virtual_id_number":805491,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 08:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-08-23 14:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-08-24 06:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-08-24 06:57:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-21 19:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 14:43:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-08-24 06:57:11.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0e2d792ac13cef72d3ec0ecc4e41f2f74e40a589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-21 18:40:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1cfa40af742d5ea01e228863f3b476a974b6fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 19:52:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88e5fa0d8cc34d9f87bf1d5714aa2445 : SUCCESS in 6m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51ad50ae258a44fa9e673e807fd5de3e : SUCCESS in 6m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2ae947cc96946ea995636ae9f0905ed : SUCCESS in 4m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3706adbef5f64c0da30f9b30565e0424 : SUCCESS in 12m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cd8f2ee1623e4cb68519a4bbf1678997 : SUCCESS in 15m 01s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/118647055d294772862152c227672bdf : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2208a58269646aca64bde1b822581cf : SUCCESS in 10m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53aaf884a8fe4b96af0811e1acad4641 : SUCCESS in 12m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f697917b1c154f35887b25868bbfe9a7 : SUCCESS in 52m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/528823e5341b4f298cb64694ba44fcf3 : SUCCESS in 1h 06m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5083b42908d44d16844737b5eff620c6 : SUCCESS in 1h 03m 55s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/c2775117419b4355b997efba2a2ab6ce : SUCCESS in 47m 39s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/656d41d3aaa244e1a49b43087b276744 : SUCCESS in 46m 21s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/777d7a3656254a439647f481ad312d4e : SUCCESS in 46m 59s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8f735873956433ba9e476c9dbbedda3 : SUCCESS in 33m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fe598fde2f1341c6b72a96bcde3d8959 : SUCCESS in 33m 27s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6efa42e1cd94449b99aa7d91694a3050 : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/47d97f014428437eb40c67a96bef58f0 : FAILURE in 11m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3f17cc0ae99e4bb6a4cb8fa676542860 : FAILURE in 9m 35s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca6632597b6b4aca9423a97dde38d04b : SUCCESS in 40m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bd53b2d289f0410697fb2047c3e47e8e : SUCCESS in 9m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b7a1049bf0c47dfbef5d88d0dc53578 : SUCCESS in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfe4c33f0eed05ee1a8ddc3a85e7d8db812d03f9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-23 14:43:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec1eb12478dfc8e97c941082b01f09657d5bbcc4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-24 06:57:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5e6659698d680bc40c049381b8b505dc8f6e979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 06:58:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0125cd89ea6e46a2633e439c38590c0380aa2d10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:59:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/274044abd84b46ccad0f135e772581ef : SUCCESS in 6m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/842ba70df09248808d24b25d2150b9c7 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72e54a81aead4eefae25d8939839a865 : SUCCESS in 6m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ae7e3d34e2364c21b7132bbaeea0d18b : SUCCESS in 17m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5013679ee08e4198a016c48ab550adf9 : SUCCESS in 15m 11s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/79987a5b43e242d78fd4804a6eb98d4f : SUCCESS in 1h 00m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f5d1f577c43d4b4d9900f0c0032739f6 : SUCCESS in 1h 09m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d3e21c2413ff4eb69f02fa3c01d263aa : SUCCESS in 1h 09m 25s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/90a1edab3d94423daf01a85087a982df : SUCCESS in 59m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a083253b6f1e4ef49c41a70c2d0a204e : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3dcf303ac3c5dfcd7be8e19dc874873d8f68c72f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:59:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4f87e89cd5376343cb35c00366629abfa28325a","revisions":{"b4f87e89cd5376343cb35c00366629abfa28325a":{"kind":"REWORK","_number":1,"created":"2021-08-21 18:40:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/805491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-oslo","ref":"refs/changes/91/805491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/91/805491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/91/805491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-oslo refs/changes/91/805491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-oslo refs/changes/91/805491/1"}}},"commit":{"parents":[{"commit":"78c768e61643f86db4031f3e03eb26ea24d61bb9","subject":"Cors: Accept array for ListOpts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/78c768e61643f86db4031f3e03eb26ea24d61bb9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 00:44:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-21 18:40:22.000000000","tz":540},"subject":"Fix the wrong default value of idle_timeout","message":"Fix the wrong default value of idle_timeout\n\nCurrently we always get the following warnings because of the wrong\ndefault value of idle_timeout.\n\nThe idle_timeout parameter is deprecated.\nPlease use connection_recycle_time instead.\n\nThis change fixes the default and ensures that the warning message\nis shown only when the deprecated parameter is set.\n\nConflicts:\n\tmanifests/db.pp\n\nChange-Id: Iba93ffdacf45096a6db65f3f0dfcfaf7d87574f2\n(cherry picked from commit 12e2f4a886f7a97d1f34bce22254ea114f829bbd)\n(cherry picked from commit 69263d76c1e9d925b96a36beba521dde756863e8)\n(cherry picked from commit 06766e8789f411e6e135f041a6b264bb70d48ab8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b4f87e89cd5376343cb35c00366629abfa28325a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-oslo/commit/b4f87e89cd5376343cb35c00366629abfa28325a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
